)]}'
{"id":"openstack%2Frequirements~739014","triplet_id":"openstack%2Frequirements~master~I575b18220db4b0a2ce677f99df86486d3ab0f150","project":"openstack/requirements","branch":"master","topic":"pbr-fixes","hashtags":[],"change_id":"I575b18220db4b0a2ce677f99df86486d3ab0f150","subject":"Revert \"Remove six from local requirements\"","status":"MERGED","created":"2020-07-02 10:38:03.000000000","updated":"2020-07-02 17:17:42.000000000","submitted":"2020-07-02 17:15:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739014-1593710119610-447c4ef9","meta_rev_id":"f917d459ace14fc4bc7f50142ce759908bf345a6","_number":739014,"virtual_id_number":739014,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-02 12:53:20.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 17:15:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-02 13:43:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-02 13:43:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 13:43:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-02 17:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f10521fb18bbdb4da87fe620212ec05634278ad0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-02 10:38:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb0a46cb16abfd6278dcdf9b06867059798a7cff","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-02 11:27:10.000000000","message":"Patch Set 1:\n\nI\u0027m confused. This is the local requirements for installing and using the requirements repo code, not the global requirements for other repos to use. What failure are you seeing?","accounts_in_message":[],"_revision_number":1},{"id":"257fd75506c7f7deea7a09c7dd77b5ff562959aa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-02 12:53:20.000000000","message":"Patch Set 1:\n\n\u003e I\u0027m confused. This is the local requirements for installing and\n \u003e using the requirements repo code, not the global requirements for\n \u003e other repos to use. What failure are you seeing?\n\nI\u0027m trying to get pbr passing and it\u0027s failing on six not being installed [1][2]\n\n[1] https://review.opendev.org/#/c/719964/\n[2] https://zuul.opendev.org/t/openstack/build/f4b5cbb048e64e24bba911bf29e3cf3a","accounts_in_message":[],"_revision_number":1},{"id":"f991b28b79127f6914d3a5470bf887a944e5ae96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 13:25:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db8c535460c34a87be5a590fcef5f560 : SUCCESS in 9m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2976cd14f254a139997c921ad4e3397 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00be677b784b432bafa5189b90308557 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feb7363d505044c0b1cf336a6a7e83c3 : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b536faa077d54f5db70169d11b403386 : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae7f05d38fcd4014925e8cb9514a493d : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0e5b5bc96354d193b986055edb4a090db14c7528","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-02 13:43:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSome additional cleanup apparently needed to go along with this.\n\nI do think pbr needs to be updated to not use update.py. We don\u0027t sync requirements to repo\u0027s local requirements.txt files, so all of that should go away.","accounts_in_message":[],"_revision_number":1},{"id":"893e987ae2e3163edd49b0c0358682ade1f72dce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 13:43:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3414b04596c20fb5dec03392d2b016ffae4b8a62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:15:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76656c1c9e3483fa8c9494b924b5aff : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19acb48530444fec83bffcc8e7e12f20 : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2ca1efaad3b435bab1957063cfd6ade : SUCCESS in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c29fc8826ebe4826be32502e22c5f64c : SUCCESS in 9m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/caff034be87e486a90909e6ad83b8eb1 : SUCCESS in 9m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/912dc190c79f4c2f953bad28350600e9 : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9d5c9fd7b1516ca2ec772c45d31694e41fc027dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:15:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4fce1dcc2b36166df214d76f453ead7f303baa4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:17:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f48a6dd263f4dcd92176ad6194e8d88 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22c362cf6488e9967a3041328daf1395f25ea8fa","revisions":{"22c362cf6488e9967a3041328daf1395f25ea8fa":{"kind":"REWORK","_number":1,"created":"2020-07-02 10:38:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/739014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/739014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/739014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/739014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/739014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/739014/1"}}},"commit":{"parents":[{"commit":"8987d385d781bf01d7caec61fa4cbd0407582e40","subject":"Bump upper constraint for oslo.log to 4.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8987d385d781bf01d7caec61fa4cbd0407582e40"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-02 10:35:01.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-07-02 10:37:39.000000000","tz":60},"subject":"Revert \"Remove six from local requirements\"","message":"Revert \"Remove six from local requirements\"\n\nThis reverts commit 9ee3854a500af72efd5d2147745d02e2fba2dd1a.\n\nThere are still a couple of tools using this, including things that pbr\ndepends on [1]\n\n[1] https://opendev.org/openstack/pbr/src/commit/9257b6843568052be1dc1a5f3df6f629795e08b2/pbr/tests/test_integration.py#L80-L83\n\nChange-Id: I575b18220db4b0a2ce677f99df86486d3ab0f150\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22c362cf6488e9967a3041328daf1395f25ea8fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22c362cf6488e9967a3041328daf1395f25ea8fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
