)]}'
{"id":"openstack%2Frequirements~739290","triplet_id":"openstack%2Frequirements~master~Ic9ee4d5164b144638c3ea3d4ddc1e6bac86378c0","project":"openstack/requirements","branch":"master","topic":"bug/1886237","hashtags":[],"change_id":"Ic9ee4d5164b144638c3ea3d4ddc1e6bac86378c0","subject":"Temporarily cap setuptools\u003c48.0.0","status":"MERGED","created":"2020-07-03 22:26:00.000000000","updated":"2020-07-04 01:23:26.000000000","submitted":"2020-07-04 01:06:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739290-1593824768151-fcaafab0","meta_rev_id":"5481950e27cdc68384cf04c01332518482399ceb","_number":739290,"virtual_id_number":739290,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-07-03 22:36:12.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-04 01:06:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-03 22:36:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-07-03 22:36:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-03 22:36:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-04 01:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2527c1aac9e5b60b22c29e00bbcb558431bb466a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-03 22:26:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bfe733c7ffe6507cb2fa5fb8e00297b681f0d12","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-03 22:32:19.000000000","message":"Patch Set 1:\n\nWas a bug filed upstream to confirm the behavior?  What is the plan with dealing with the new paths going forward (aka, when can this patch be reverted)?","accounts_in_message":[],"_revision_number":1},{"id":"b759fa582874798f45dfc96c6c3415a423fe87f8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-03 22:36:12.000000000","message":"Patch Set 1:\n\n\u003e Was a bug filed upstream to confirm the behavior?  What is the plan\n \u003e with dealing with the new paths going forward (aka, when can this\n \u003e patch be reverted)?\n\nAs of now we are not sure if it is setuptools bug and they can fix it or we need to update devstack for the \u0027pip install -e\u0027.\n\nFor now, I have filled devstack (critical) bug for further debugging.  https://bugs.launchpad.net/devstack/+bug/1886237\n\nAs part of that bug (either we fix in devstack or setuptools fix it), we can revert this.","accounts_in_message":[],"_revision_number":1},{"id":"c5742aa9de11a0a696afb8765f3895ebc72dec04","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-03 22:36:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5bad9b1017941f3477227c836fac074d0fcd2d8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-03 22:36:47.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c049396fb7326dc448e8d8098cb09eaa0a41347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 23:39:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808d61e117134763b7069361096549d3 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/100d5b6434a14376a5a2d45a52da2177 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a4131e09379458c8c329a0e22716b27 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cb325a5892e4674aa9ea98efd4351be : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/248115621f0745ae87cef58c5658e101 : SUCCESS in 5m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95f614e05784427cb89fb66c5d8c0a4c : SUCCESS in 47m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df96977bba7a41dd99437ea432bd0113 : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f787a6c0abbc8986edca2ee92d22003246097b27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-03 23:40:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8dcb418d8a9cc89f01876746c7f86b97355407e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 01:06:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c46e0e049dc4f84812d58242584c6ec : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb5300b07a445ebab8cf21cffad75bf : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a5b9b12e16046e88982467352f0f6f2 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47614c01401c48c88c8272dda2d42b8c : SUCCESS in 4m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d323bc354100489ba6bc618a9bd02266 : SUCCESS in 3m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7ead9ef735b4997a9ca839245a0f141 : SUCCESS in 51m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffe24bd1be174d94adffb5830189f45f : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"57c8d18a093913690e06a6569b146adfcce8f937","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 01:06:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c35d78968ab166344a0dcc68ea20768bd6001c8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 01:07:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90adbf637daa43a4bfd2f7a953440438 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1},{"id":"86879f81cf3dcd21f0b855f129395ade268c726c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-04 01:23:26.000000000","message":"Created a revert of this change as Ib71d06242d94aa1a73e69b0c7423b973ae685a1f","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca1eb8dd0ae9c7c7ac149e8b4ff9bbc9630eb43f","revisions":{"ca1eb8dd0ae9c7c7ac149e8b4ff9bbc9630eb43f":{"kind":"REWORK","_number":1,"created":"2020-07-03 22:26:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/90/739290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/739290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/739290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/739290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/739290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/739290/1"}}},"commit":{"parents":[{"commit":"22c362cf6488e9967a3041328daf1395f25ea8fa","subject":"Revert \"Remove six from local requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22c362cf6488e9967a3041328daf1395f25ea8fa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-03 22:22:42.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-03 22:22:42.000000000","tz":-300},"subject":"Temporarily cap setuptools\u003c48.0.0","message":"Temporarily cap setuptools\u003c48.0.0\n\nwith new setuptools 48.0.0, devstack started failing on\n - usr/local/bin/keystone-manage: No such file or directory\"\n\nhttps://zuul.opendev.org/t/openstack/build/55e53d1c37194d8992f79a4090da86ee/log/job-output.txt#10572\n\nIt is blocking all the gate for now. We are not sure if that is bug on\nsetuptools side or it is incompatible change and we need to update devstack\nfor that but till we figure out that let\u0027s use the old setuptools to unblock the gate.\n\nChange-Id: Ic9ee4d5164b144638c3ea3d4ddc1e6bac86378c0\nML: http://lists.openstack.org/pipermail/openstack-discuss/2020-July/015779.html\nRelated-Bug: #1886237\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca1eb8dd0ae9c7c7ac149e8b4ff9bbc9630eb43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca1eb8dd0ae9c7c7ac149e8b4ff9bbc9630eb43f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
