)]}'
{"id":"openstack%2Frequirements~739330","triplet_id":"openstack%2Frequirements~master~Idf9ae2ea576f4ed882489f03e4dbc94193243c98","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Idf9ae2ea576f4ed882489f03e4dbc94193243c98","subject":"manual generate-constraints run","status":"MERGED","created":"2020-07-04 19:05:56.000000000","updated":"2020-07-05 04:22:29.000000000","submitted":"2020-07-05 04:21:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739330-1593922873585-f3086ce0","meta_rev_id":"d317c9b7d0d1922844f1bd4b13236e931073fef1","_number":739330,"virtual_id_number":739330,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-05 04:21:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-05 03:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-04 22:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-05 03:25:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-05 03:25:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-05 04:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3f7cb3e0e14a51843cf036681c62788ba622e57","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-04 19:05:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d075178c1d9b7e8fcd2fd42f7c0716bd62d952c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 19:51:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/29bbc8dbb9c74330a1f2686c795f016c : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"168b7a6cc95e4477e3470d8af220aba8ac8edd27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 20:26:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3d46113a2a54c14b9d25f922f0772fc : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd4475d4dc34fd2978778ba2ac6540b : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a36764a963a4f7aa3d3237dae851b2a : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de5c55138ef84fe89218a7a162ca39de : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f9cedd93c2c4d829569c68a4f2eb6bb : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eb5ac17e5f9f47c7894ce2733bc971e0 : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/62f61e09daf74c6fb3736077d41d54e2 : SUCCESS in 8m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f163614806e746cd9411ccab6f48e801 : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a171d3352afd4aff8cf922332f5133a6 : SUCCESS in 8m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/216d3e8a3099449ca7b79fc063f30267 : SUCCESS in 4m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3c74266811744e5a9c8aac46f4e2545b : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8001c05c909c473e9731094aef991c44 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ff068dc014be42aabde51f95c8c9828e : FAILURE in 5m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a0594afd3b5b41818d4ecb6fc208894d : SUCCESS in 13m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ecac11cf40d44d4c9cab4c451ce769b0 : SUCCESS in 25m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/843ec446c2a24a2b9e86a569e0c2e49d : FAILURE in 11m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ad6d1daa9d24db5a11814a4a4447e6b : SUCCESS in 18m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/76a8e79dcd2e4eebae92441854822680 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/81a0cbbb7b6743a4a59fe2ab67957620 : SUCCESS in 6m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/13a6b3d89b4241eb9c431d901ecd0d5d : SUCCESS in 9m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbab3cd8b32c486d8e166209e15623b9 : SUCCESS in 44m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/780d31ada3da4aecac2362f59548aadc : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"063046e9c210ba86d554d4d499f6b89cfae9dca1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-04 20:32:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4791fb5bfb6586cb90153b1b83d3c63444c9806f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 21:12:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ac1ce2db2c1544b2ae7d3f24e651e164 : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":2},{"id":"997eca8331583feefa5be90454f86ac06bf35d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 22:07:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/332e47e2407c4c5bab5a70df03e713cb : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a665ec03ca94be09a0f893b707e8261 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e656b5781a5f4b998d77ed770dd5d72b : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10b7d6c5aafd4c938fd0dd60c9186492 : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a391d1f0ebb644f3b86ea37541f4bcdb : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ac39e048d7a4602ad805203ea2ac0d0 : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1e9e0ac2ca1f4cdfb6b9760267dc0df5 : SUCCESS in 10m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb63a720e165492fa9fa049f713523dd : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/004bff21e9eb4ed687f8e3e3a4f4991c : SUCCESS in 9m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d785375150954166a0975c3cedd7298c : SUCCESS in 4m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5b99569725242b4a15282b0ffed8469 : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1fcae43c92c45ef8e1c833b8fb565ef : SUCCESS in 7m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bbcadab9d76b47b4aca140e5bd14c3bd : SUCCESS in 12m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ebbaee45f478471daf8ab0eccae6f5cb : SUCCESS in 24m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cf84714a22484039a888e605baa0b76d : SUCCESS in 25m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ad3a295ca2f844c28fc31a18c7a2f9af : SUCCESS in 15m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b9e17934711d40c5a6a9062034f8809d : SUCCESS in 16m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a06b9e2bb35f4fbd93af4672672a6d0f : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2500f844bc174fd9bbb06fbc2ec45495 : SUCCESS in 6m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7ffa69d57bfb4fa48763833ab212adb5 : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e3a50b950334e1ca3afa4a580ec3883 : SUCCESS in 55m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2ab11586619479182765342cd37c7d4 : SUCCESS in 1h 33m 18s","accounts_in_message":[],"_revision_number":2},{"id":"0b4b3d8c14fbcf46f75cacf5c33a656d35846cfe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-04 22:28:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"22967ef8147ef6c26700fdf797bcea96c761231f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-05 03:25:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"237ab209dbdf41fa19374a650af1916cecb7f00f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 03:25:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"044d84606e5a133c7c807b3453595c68bab528cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 04:21:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b640548e9dc04a45a882a032cb83cf13 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7af29fd858e24267af209b3ba67d7956 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cac56b2ddaa490caf47a69a909634ef : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/942c2560d0c642169cc0a3bc9751972e : SUCCESS in 4m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1241989b5e8542059501c38421688eb0 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eba61af8e0e94717bae11358421b90cc : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d8f4f1e4094f4727af768d240a6f2733 : SUCCESS in 9m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a1ab121a18074c8a9d351061b2019aa1 : SUCCESS in 8m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7ecb22fce7bb4d93be4fbe4454c02ebf : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/32ef09d6e7cb4dbe87a224ce21b7e69b : SUCCESS in 4m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/55b780b40ea94e2b9a619915eb69fb60 : SUCCESS in 6m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d7f4c19c57c45f287715424f11707d6 : SUCCESS in 8m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/58c249901cd940dc822fa6c549ad0a51 : SUCCESS in 6m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/68dc24fe97c44a9fb315bd89dc28572a : SUCCESS in 19m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b6b50519c3d4720997f8284cea0ff59 : SUCCESS in 23m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cd1044ae4e9e48c796620ebc7640faaf : SUCCESS in 9m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/04031b98f31f41d689e9e43506f9956b : SUCCESS in 15m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3eb5892b9abe456794b38e15140622d8 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/547d3805304d4d75899fc72c83a901f9 : SUCCESS in 6m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a7f5a7b9e49d4d909d0db5c2814cd1ad : SUCCESS in 11m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/71d87ee2e3c346b0a654ce0e4d3c1d62 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65874aedae084ce1ad88951d1ebc3730 : SUCCESS in 50m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ece708af33c48b08c634097acb48f22 : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":2},{"id":"6fd13415663efaf65552c65793734722335d8d9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 04:21:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d317c9b7d0d1922844f1bd4b13236e931073fef1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 04:22:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b750cf520c4c43b7bff50dfbe28fb6 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bdb9d64daf8c667f5dfb043fcfca5670ba485a73","revisions":{"27850bda17c727d0c6218e1970718101e301d644":{"kind":"REWORK","_number":1,"created":"2020-07-04 19:05:56.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/30/739330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/739330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/739330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/739330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/739330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/739330/1"}}},"commit":{"parents":[{"commit":"ca1eb8dd0ae9c7c7ac149e8b4ff9bbc9630eb43f","subject":"Temporarily cap setuptools\u003c48.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca1eb8dd0ae9c7c7ac149e8b4ff9bbc9630eb43f"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-04 19:03:44.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-04 19:03:44.000000000","tz":-300},"subject":"manual generate-constraints run","message":"manual generate-constraints run\n\ngenerated via:\n\ngenerate-constraints -b blacklist.txt -p /usr/bin/python3.6 -p \\\n  /usr/bin/python3.8 -r global-requirements.txt \u003e upper-constraints.txt\n\nheld back the normal mock, msgpack (because of salt) and docutils\n(because of boto iirc).  Have not held back oslotest (mock issues) or\njsonpatch (ironic issues).\n\nChange-Id: Idf9ae2ea576f4ed882489f03e4dbc94193243c98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27850bda17c727d0c6218e1970718101e301d644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/27850bda17c727d0c6218e1970718101e301d644"}]},"branch":"refs/heads/master"},"bdb9d64daf8c667f5dfb043fcfca5670ba485a73":{"kind":"REWORK","_number":2,"created":"2020-07-04 20:32:55.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/30/739330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/739330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/739330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/739330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/739330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/739330/2"}}},"commit":{"parents":[{"commit":"ca1eb8dd0ae9c7c7ac149e8b4ff9bbc9630eb43f","subject":"Temporarily cap setuptools\u003c48.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca1eb8dd0ae9c7c7ac149e8b4ff9bbc9630eb43f"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-04 19:03:44.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-04 20:31:55.000000000","tz":-300},"subject":"manual generate-constraints run","message":"manual generate-constraints run\n\ngenerated via:\n\ngenerate-constraints -b blacklist.txt -p /usr/bin/python3.6 -p \\\n  /usr/bin/python3.8 -r global-requirements.txt \u003e upper-constraints.txt\n\nheld back the normal mock, msgpack (because of salt) and docutils\n(because of boto iirc).\nheld back oslotest for mock issues in nova.\nheld back jsonatch for issues in ironic.\n\nChange-Id: Idf9ae2ea576f4ed882489f03e4dbc94193243c98\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdb9d64daf8c667f5dfb043fcfca5670ba485a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bdb9d64daf8c667f5dfb043fcfca5670ba485a73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
