)]}'
{"id":"openstack%2Frequirements~739912","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I158bcc8c249ad7ec981707c0d96ad8db87eb23d9","project":"openstack/requirements","branch":"stable/ussuri","topic":"bug/1886422-stable/ussuri","hashtags":[],"change_id":"I158bcc8c249ad7ec981707c0d96ad8db87eb23d9","subject":"Add \"isort\" library to blacklist","status":"MERGED","created":"2020-07-08 07:51:29.000000000","updated":"2020-07-10 13:23:14.000000000","submitted":"2020-07-10 13:23:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739912-1594387394448-167245bb","meta_rev_id":"e62e90084dd05328773a9502196948b595889bc7","_number":739912,"virtual_id_number":739912,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 13:23:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-07-10 08:58:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-10 09:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2020-07-10 08:58:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-07-10 08:58:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-07-10 08:58:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-07-10 08:58:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-07-10 09:50:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 08:03:20.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-08 10:08:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-08 10:49:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-07-08 11:08:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-07-08 11:13:09.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-07-10 01:23:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-10 09:50:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-07-10 13:23:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0957c9170f59500a67f653ec8aec9592322d350e","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-07-08 07:51:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fb6407edc46bcb5773183a7859c5896e2886049e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-08 08:03:20.000000000","message":"Patch Set 1: Code-Review+1\n\nNeeded for https://review.opendev.org/#/c/739861/1 to pass","accounts_in_message":[],"_revision_number":1},{"id":"a275d2a5a4971dd4bc054f7035379944e25761ec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 10:08:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bd2ca1bac06809c0cbbf0ef6e1a83d8e2015dfd4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-08 10:49:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a2c7a162355120363232b09fdd175032a80fc521","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-08 11:08:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"422a21496840746ffe2e613fd222529ec135224f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-07-08 11:13:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ccdd7bfdb403928334c1ecae1a442cd5a1029b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 11:24:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f0cb01705a3407180c89ff449762cc2 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51376c8860b446f18195aeb646fc863b : FAILURE in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3e0031cabac45828d1f4ee39ede0471 : FAILURE in 8m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5c88a36eeb44f238deafde015a14bba : SUCCESS in 7m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebefa188800e4224b56236bd7b810da1 : SUCCESS in 1h 02m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/351b7b56a5f44a82918408ab8e2950b4 : SUCCESS in 1h 32m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ca96313f30c4a1bb01a9306c23a6c3c : SUCCESS in 1h 28m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/496a71d0833e4e52851d3b1d05180352 : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2f4f06835e051592d7fb9f152cbaf899f485b00c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-08 12:16:35.000000000","message":"Patch Set 1:\n\nrecheck parsing test errors","accounts_in_message":[],"_revision_number":1},{"id":"e64087e3f20722c015725d3a732b354177cce7e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 20:34:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1dc836e24614c2a8edf92d16bcfb820 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15ba18c604ff4a9db502e739640f4efa : FAILURE in 10m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8af33cbe79d4828aa75f7c288e4c9f7 : FAILURE in 8m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e69965457fc493785c9ecb045dcfb8d : SUCCESS in 5m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb659c3650c84f15bbf4e44be5fdd486 : SUCCESS in 54m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/41eaba9e8c564b0887af30ef27b3c4d6 : SUCCESS in 1h 18m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f76c80d5b0194a54a9336cd8bacc5e9f : SUCCESS in 1h 08m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/0ea05e1aa2eb499b8090381806994bc9 : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0299b9ac77af9e317cce9376786edfcdfbadf961","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-07-09 12:44:20.000000000","message":"Patch Set 1:\n\nrecheck holding a node, failed to reproduce locally","accounts_in_message":[],"_revision_number":1},{"id":"a22e8851a05a57c082f1b603eeb1958fca595d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 18:42:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5764b6a0f67544fca820f228f70a012d : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17844547c02e4f1a926c0d127d4327e7 : FAILURE in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/640e0fc76b75415087a7de073b4f0812 : FAILURE in 7m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0639173b8f24c71a74e29116ede9912 : SUCCESS in 7m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f999b190fa74714b4d21f0099cf9699 : SUCCESS in 54m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f4314c5634347b3bb3f85285725954f : SUCCESS in 1h 21m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/513e5883bfc24e9e8bd98c21b7cf836f : SUCCESS in 1h 21m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/19254bbd666a4cab927855d0f8a11184 : SUCCESS in 1h 09m 39s","accounts_in_message":[],"_revision_number":1},{"id":"dd5c10241194a3892cbd3d42a0a9a61ab2cb9283","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-10 01:23:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"12c666d8d2b6d16d3f6d236a8b31ed08dd0a956c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 02:58:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf07df3f7b2448e84cb07b3765b0b14 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/833aa817e43344c2bfe3784043b8286e : FAILURE in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bc80b82c392425db64e123ccfcf0b3a : FAILURE in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/adc00d2499bb4f2b91d6ff60befa1f45 : SUCCESS in 5m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8beb29a9f01a462db2a5716b1c59d672 : SUCCESS in 1h 02m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b83abc5b473a406bb3f975e70d4e0b7e : SUCCESS in 1h 08m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e5dca452c374b14921954422cb7f235 : SUCCESS in 1h 28m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/a618f04cefd141398322c8096805c008 : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9836556d46b6c5dab30ad6189ce0f75a5fa8b2d3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-10 08:58:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fbf797da9b70a7f085ac31d8372703c7bc7dfd22","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-07-10 09:50:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d49ebbc1b1eceee2ffafd769cf38983bbc3c39a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 11:01:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d355637039b042a1b5c081595e6e0aa8 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d560be2029d4445906016c68f2f8ff3 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ca6545491124384ae56d2923eb1ad69 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b75048020d042828f1868177f5a9623 : SUCCESS in 6m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6252a5a07cc54c50a7b35ae2c3bc2a8a : SUCCESS in 1h 01m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3b5511ec9514602915d702c66f66fcf : SUCCESS in 1h 36m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d37e881618754f7bae8c2c4338af1500 : SUCCESS in 1h 24m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/360c0cd4af984f9f9858f18e38b8bde9 : SUCCESS in 1h 20m 12s","accounts_in_message":[],"_revision_number":2},{"id":"98bbc3680f946ec42fdc88e8af5d6ee7c6126991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 11:01:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"846d70eb73e585e03d94b4851c910ddebf19e890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 13:23:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6206e03cbf404d9d9012d7be3ea78b16 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/095dff633bf743e98f41bdfb65fff77e : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adacdf80ff5e4c1faf9288460e7afb84 : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98809ef1448c45fea36b0ed1635491ee : SUCCESS in 4m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/964ef06ea5da4b81bee966c8c72118e8 : SUCCESS in 50m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb20cdf55450400f87639bb4cea249cb : SUCCESS in 1h 14m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/146e22a86f494b629af4e83a5ee7f61c : SUCCESS in 1h 02m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/38cb959e803d4f55a3c0cfed26209993 : SUCCESS in 1h 13m 13s","accounts_in_message":[],"_revision_number":2},{"id":"8bfcc741470e7caa6fc38d78003673dcec7fbdb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 13:23:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"59385b481e904e4f3e33f85564f3ffece96b655d","revisions":{"36afc8aaa18662960e0bec24a707429a5e180078":{"kind":"REWORK","_number":1,"created":"2020-07-08 07:51:29.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/12/739912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/739912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/739912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/739912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/739912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/739912/1"}}},"commit":{"parents":[{"commit":"42daf674222e85fb4dc3d4d1bb113389ca3a7137","subject":"Merge \"update constraint for oslo.policy to new release 3.1.1\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42daf674222e85fb4dc3d4d1bb113389ca3a7137"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-06 12:24:30.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-07-08 07:51:29.000000000","tz":0},"subject":"Add \"isort\" library to blacklist","message":"Add \"isort\" library to blacklist\n\n\"pylint\" requires a class named \"SortImports\", not present in\n\"isort\" versions \u003e\u003d5.0.0. Until \"pylint\" fixes this problem, we\nshould allow each projecty to install its own \"isort\" library\nversion to unblock CI pep8 jobs.\n\nChange-Id: I158bcc8c249ad7ec981707c0d96ad8db87eb23d9\nRelated-Bug: #1886422\n(cherry picked from commit be1dca06e04a68e6ad02f293369124d3c24875ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36afc8aaa18662960e0bec24a707429a5e180078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36afc8aaa18662960e0bec24a707429a5e180078"}]},"branch":"refs/heads/stable/ussuri"},"59385b481e904e4f3e33f85564f3ffece96b655d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-10 08:58:30.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/12/739912/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/739912/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/739912/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/739912/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/739912/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/739912/2"}}},"commit":{"parents":[{"commit":"e04d3ada5af0e689fb03719eac7bc55a32ae646a","subject":"Fix failure with setuptools \u003e\u003d 49.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e04d3ada5af0e689fb03719eac7bc55a32ae646a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-06 12:24:30.000000000","tz":0},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-07-10 08:58:30.000000000","tz":0},"subject":"Add \"isort\" library to blacklist","message":"Add \"isort\" library to blacklist\n\n\"pylint\" requires a class named \"SortImports\", not present in\n\"isort\" versions \u003e\u003d5.0.0. Until \"pylint\" fixes this problem, we\nshould allow each projecty to install its own \"isort\" library\nversion to unblock CI pep8 jobs.\n\nChange-Id: I158bcc8c249ad7ec981707c0d96ad8db87eb23d9\nRelated-Bug: #1886422\n(cherry picked from commit be1dca06e04a68e6ad02f293369124d3c24875ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59385b481e904e4f3e33f85564f3ffece96b655d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59385b481e904e4f3e33f85564f3ffece96b655d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
