)]}'
{"id":"openstack%2Frequirements~739914","triplet_id":"openstack%2Frequirements~stable%2Fstein~I158bcc8c249ad7ec981707c0d96ad8db87eb23d9","project":"openstack/requirements","branch":"stable/stein","topic":"bug/1886422-stable/stein","hashtags":[],"change_id":"I158bcc8c249ad7ec981707c0d96ad8db87eb23d9","subject":"Add \"isort\" library to blacklist","status":"MERGED","created":"2020-07-08 07:54:12.000000000","updated":"2020-07-10 15:44:49.000000000","submitted":"2020-07-10 15:44:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739914-1594395889823-2400accf","meta_rev_id":"f776a8e094cb5fc30eeb6e0f173a4755faad6b1d","_number":739914,"virtual_id_number":739914,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 15:44:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-07-10 09:08:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-07-10 13:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-10 09:08:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-07-10 09:08:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2020-07-10 09:08:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-07-10 13:49:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 08:03:42.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-08 10:08:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-08 10:49:57.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2020-07-08 11:09:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-07-10 13:49:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 15:44:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3368759c729b36940a9945557929e9c2e8fc76ec","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-07-08 07:54:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b13a7f357e21f91caa8bc6fd1fc955180a3d8bae","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-08 08:03:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cdd5a6bbe0887955eb23b20ea2be458380e59c5e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 10:08:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b01d3658fac4dd76eff082e8f66107ea3292fc7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-08 10:49:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"190d4003813700e22a7acfa619bbde44a98cea1d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-08 11:09:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8fd27be071bfd9859854efd33243a3b484047eb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 11:42:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20f8bc734e5c4f4b83b4eed9e7d5042b : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1ee779ff4584a4099f0ec8f62e0a20a : SUCCESS in 7m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/56e44d87aaf54f29a79d5b233703ace6 : FAILURE in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1254d4a04de349319e2d307fc3d21081 : FAILURE in 7m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d752a2fa54dd4b51b237ec492cd0091c : SUCCESS in 8m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f37ec958bdbd4303950a1c94140b8cdd : SUCCESS in 1h 18m 34s (non-voting)\n- tempest-full https://zuul.opendev.org/t/openstack/build/fb3921dc569e4f76b0e5db3716e8748e : SUCCESS in 1h 43m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91c4ec9a5a80465f8ccd170bf55d4928 : SUCCESS in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecf2ac1f97fbc790320dfec4697d2985e2364a77","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-10 09:08:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"33f1fffdeb30624e2a5602c7d4a8c7e8699d29dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 11:30:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b08362a0e6b24286bdea949d04afbf9a : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d856a85addaf4b78822e73908f4aac28 : SUCCESS in 5m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3e45a2291e674eaeba394103ed6796e0 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54072f91a773460ab18bec168ebf5ca9 : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52c39394aaad4376b261c47284cbe60e : SUCCESS in 5m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/30ea27e3238748e6b1f9b66b18fc6bef : SUCCESS in 1h 05m 44s (non-voting)\n- tempest-full https://zuul.opendev.org/t/openstack/build/8235964c41a8445cb481ff0b66148519 : SUCCESS in 1h 52m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b86210d1218e4a9fb64afab27d1f01ae : SUCCESS in 1h 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"085ba0ad472402bea174c4ec390896027a54349a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 13:49:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bf369079239ef6ab3f35c00e454a00175d39bf96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 13:50:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"679d1db5d10952f268397819b57a8fbe22b414ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:44:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8a7d0e54fcba35913e114031b487fc718a7637b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:44:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59eee6011d544253b983bdf51ca3536d : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19d518627dbb4c81ab03723b2331d08e : SUCCESS in 9m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ee2cac79ac0048d8ad1507e5e8d4fb0a : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7c0d39765db4e1caa6c93b3d56456d3 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/481af0b656f04ab887cfde5b893f2be5 : SUCCESS in 4m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f6694f3435604b979029ad03b88da775 : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8489fb44bccf16640941114854f70c6df99b8783","revisions":{"bc20c46b7875bc3f37c76d1902a82027cf904f9c":{"kind":"REWORK","_number":1,"created":"2020-07-08 07:54:12.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/14/739914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/739914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/739914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/739914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/739914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/739914/1"}}},"commit":{"parents":[{"commit":"d1b88cf8fe082f68de2837389f9f1a1279586b00","subject":"Bump subunit to 1.4.0 in train too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1b88cf8fe082f68de2837389f9f1a1279586b00"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-06 12:24:30.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-07-08 07:54:12.000000000","tz":0},"subject":"Add \"isort\" library to blacklist","message":"Add \"isort\" library to blacklist\n\n\"pylint\" requires a class named \"SortImports\", not present in\n\"isort\" versions \u003e\u003d5.0.0. Until \"pylint\" fixes this problem, we\nshould allow each projecty to install its own \"isort\" library\nversion to unblock CI pep8 jobs.\n\nChange-Id: I158bcc8c249ad7ec981707c0d96ad8db87eb23d9\nRelated-Bug: #1886422\n(cherry picked from commit be1dca06e04a68e6ad02f293369124d3c24875ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc20c46b7875bc3f37c76d1902a82027cf904f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc20c46b7875bc3f37c76d1902a82027cf904f9c"}]},"branch":"refs/heads/stable/stein"},"8489fb44bccf16640941114854f70c6df99b8783":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-10 09:08:08.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/14/739914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/739914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/739914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/739914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/739914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/739914/2"}}},"commit":{"parents":[{"commit":"5b1786e4a8dbac1ae7c5ffd5511164e2441cba19","subject":"Fix failure with setuptools \u003e\u003d 49.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b1786e4a8dbac1ae7c5ffd5511164e2441cba19"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-07-06 12:24:30.000000000","tz":0},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-07-10 09:08:08.000000000","tz":0},"subject":"Add \"isort\" library to blacklist","message":"Add \"isort\" library to blacklist\n\n\"pylint\" requires a class named \"SortImports\", not present in\n\"isort\" versions \u003e\u003d5.0.0. Until \"pylint\" fixes this problem, we\nshould allow each projecty to install its own \"isort\" library\nversion to unblock CI pep8 jobs.\n\nChange-Id: I158bcc8c249ad7ec981707c0d96ad8db87eb23d9\nRelated-Bug: #1886422\n(cherry picked from commit be1dca06e04a68e6ad02f293369124d3c24875ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8489fb44bccf16640941114854f70c6df99b8783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8489fb44bccf16640941114854f70c6df99b8783"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
