)]}'
{"id":"openstack%2Frequirements~740089","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ie39ebcb919e0b56ee7ea04f97a4f317ff010d030","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ie39ebcb919e0b56ee7ea04f97a4f317ff010d030","subject":"update constraint for oslo.policy to new release 2.3.4","status":"MERGED","created":"2020-07-08 18:57:03.000000000","updated":"2020-07-12 10:23:28.000000000","submitted":"2020-07-12 10:23:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740089-1594549408848-1939c49e","meta_rev_id":"b86897f9544f61397911b0ef94f961f461f47a78","_number":740089,"virtual_id_number":740089,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-12 10:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-08 19:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-11 17:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-10 09:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-11 17:53:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-07-10 09:59:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 19:00:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 09:59:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-11 17:53:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-12 10:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b96188edac6fcba4d53781a09fed9f05aea9b48c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-08 18:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a3c5c752184633571796255a509f3f1963a8d77","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-08 19:00:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f96e8baaf7e8bc4721b53a3a19efc50999d8335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 00:04:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd3ddace566a4f72aae96d77cf94fd6d : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/992385692cbe4a829678ead7a9e60fa5 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d69bd5122014654aabf68a72d4ee13f : FAILURE in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/133238427f69486695a0d4b38cc4abf9 : FAILURE in 7m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f93053c458a941d883f918dadc7ac779 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92391d2f71544bbc9b506dc0a6dcbfed : SUCCESS in 5m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4b8ad420895d48b2a1c0edf03af15619 : SUCCESS in 9m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ef14e7a94dcf4cdd9bddb9d114ddab18 : SUCCESS in 7m 43s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/f5c649f662724e14a349e6dda94b5755 : SUCCESS in 16m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/964b83ae1a8f409abd3571a7fcdb7445 : SUCCESS in 9m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/dc933e29f9a1497e8fc7ce793a8ee92b : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cca6b5fcb1484386af47913297707ee1 : SUCCESS in 9m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/957efc58a1ec48f190bec417701fc6d2 : SUCCESS in 12m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c2cf37bc9e7f49e19e10dab890ab148a : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/abf2f748d59d43b1ac9be442ad1d96a5 : SUCCESS in 8m 19s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ccb97bba4357458dafc419e1b61443f2 : SUCCESS in 19m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c36c7762406c499d90ff8450386b9cb7 : SUCCESS in 20m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/18d2db56d7e94e5e83ee4624acad3c41 : SUCCESS in 23m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ceed7ba56ed3463ba6b51ae56e23710f : SUCCESS in 26m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c7e346ce8b0b491db3113025ba9288b4 : SUCCESS in 14m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/de12b4c2d655413186684edbcfb5a88b : SUCCESS in 16m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9b828072d43b42fca946a23d49914f80 : SUCCESS in 15m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3278627dd8c4e48b5f1d96f6ec2f30b : SUCCESS in 20m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a02130177054c64b1b9e75b6820338e : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/263b841241ee4aacbee6f6080e8e2798 : SUCCESS in 12m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ccbe30fb2fd74dc08b1a6b3f6100c2a4 : SUCCESS in 9m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f33d6e6cb8354aa48f1140a56a59156a : SUCCESS in 1h 23m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4105c359ff6e45d0a39b7b55cc6fbbb3 : SUCCESS in 1h 29m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d5d224287ae2472cb156e06027f2ab26 : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":1},{"id":"942d8b57c407d3b5472bcbf81afae45289cc52b2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-10 09:59:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"247a3780648c9d2117ed61e070614f0d019f55db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 12:12:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d2491bd5834887a1352fdb7ee6d160 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e05e98e2c6c45199bd9516ab6a4d63b : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81f1e30c768c4c82b29779fe69e78a98 : FAILURE in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb6dc2f3015444bbb787bf386f5a9cf5 : FAILURE in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cefb65691ce495fa0626b0f8e6ecf19 : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3430f1eedb5745dfa677ed9cbc0068d5 : SUCCESS in 4m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/45bcac2407674e4ca57ca73450286f3a : SUCCESS in 12m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/df270f0fded644f99ec1642e2cae0a5a : SUCCESS in 8m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/43264c4abdc243de98186187f7e610b8 : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4c3658def5734947a9abaa3df1a852a5 : SUCCESS in 10m 06s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/892ebb17a3774b3dbf49ace1e1dffdb0 : SUCCESS in 6m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b555a3a126fe4ea696e93cec747fbeea : SUCCESS in 5m 43s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d27e868908d545de8121631c870e75ba : SUCCESS in 10m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/61468da34224416098b33353cec9397d : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/feeef906727d493ba41079a932e1fa9b : SUCCESS in 11m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c2fd0ec6ab814403a5037ae4ace1ee96 : SUCCESS in 21m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15f974b71e8e49e7a6ad3b4f276d1a80 : SUCCESS in 22m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/699a4d4f81dd49fa84c1643f3935c171 : SUCCESS in 23m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2254f6079c3e4a3ca7565bb42d344db2 : SUCCESS in 21m 49s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/91946d85ca61403f8ae86cba4364467d : SUCCESS in 21m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0efa9a82a8814a299e62ec9daa43b659 : SUCCESS in 14m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/611b8f00c5924bc0a50a1d289d7a1ffc : SUCCESS in 22m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73b3c7cc3d8e431698891dc5ca25200e : SUCCESS in 22m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33697bb739c943b299989ff50ea6bb85 : SUCCESS in 5m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b399872271a24f0cac554bc1ce8b52ba : SUCCESS in 9m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9e2752cd2efa4d86a0bef4d764206c72 : SUCCESS in 10m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfa2e5b129a04823b938fb460de5ad02 : SUCCESS in 1h 17m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c28e14d3ef3d48bc830f4df78a73d6e9 : SUCCESS in 1h 52m 15s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5c7ea0ce9d54434c962c3142a34833f8 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7fcd78618a053b5afac54bea580eaa1b30cac9c3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-11 17:53:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c25e3f3419d3fb4c5bd0e65bb4c5d13d824c5788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 19:27:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e7ef8342b443c6b69a836f501747da : SUCCESS in 8m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6381fc2de1a34aa5a42d440e84e9952e : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc4634b284dc4ac291ea4cb5b5afcea1 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b288178d179e4cbf8eb9f0dee838c4dc : SUCCESS in 9m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6dd5de7ddb104c32a6664a9d2affd50b : SUCCESS in 8m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0cfc82f16e8d40b493dc9983ff3ed76c : SUCCESS in 8m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/49f17670de0e4893bf5b1ab9e2b2d6ae : SUCCESS in 17m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d223cd2d339e4479994c3c5f23f53d19 : SUCCESS in 23m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9f57f6adad72458183073964cfc5dad6 : SUCCESS in 16m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/267725232af84c99915bfa597b57bf0b : SUCCESS in 25m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/99dfb54682be440fbce5b4c413ea45df : SUCCESS in 8m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f986d7fe7f5043469e4a39b77d9ac4d2 : SUCCESS in 10m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0e8500f13dcb4353921efd36497bab72 : SUCCESS in 12m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2e78937faef74e4e9ee1089d1d1891d2 : SUCCESS in 13m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ba80e247ef843198fd00740221b2565 : SUCCESS in 11m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9c264235563940249b221d5bb4355fc4 : SUCCESS in 44m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/12c6457ed9f8423e995947183a507073 : SUCCESS in 33m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/cf1d0cbaff1e4cb0bb8799a26e61c3cd : SUCCESS in 23m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/26ed418c67d24f108d47b0c767b9a431 : TIMED_OUT in 41m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/461eb6236d8546a5a93b92c587c58c90 : SUCCESS in 25m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c3a74876c19044d786449c1c1762132a : SUCCESS in 36m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7d990faff1ab40f3905244bda60913d8 : SUCCESS in 26m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75e6b938cb724573ae8d7a0852223ce7 : TIMED_OUT in 41m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f52f6934dc644669ae1c4a84d08594a0 : SUCCESS in 5m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6a6946903e944308a704aa9b670da60d : SUCCESS in 9m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e4442ee83f4441af9d511cf5c03a1ab0 : SUCCESS in 8m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5bad0d796ea7435fac70fd4169b20913 : SUCCESS in 1h 13m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/59de5920bb29454fa390ce4dc13df980 : SUCCESS in 1h 28m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dee1b604298c41f78834589423f949a4 : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"05b7eed8cb2857434bd8c389bc95c3b5b84c9d81","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-11 21:02:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b7533bddad892da13bec86e6cc16a286bf21502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 22:17:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bc0db1efbf24c509e27d08ec2fb866f : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a21cd5bc8e3429fb45fa2e82ab7a682 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be53f03b195c4fab9df5e6d905c9d5d3 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbce339e6aa94950ae86d850c3118152 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a25f694154c448c6b49176c2fdc8db8a : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12cbaa82bee241b99aa2eb866232df9a : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/712c652f240b4766968c361e721e8a25 : SUCCESS in 11m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8d4600fe75d44e0fbb4f79cd3f55083d : SUCCESS in 6m 55s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8edd322c2f7d410eb65b5cec87c2ef57 : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5036e72a31174dbc87a0696810f008b5 : SUCCESS in 8m 51s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ae7bd61a2df543bd99a6c36a4e335e4b : SUCCESS in 4m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a09953716d8b423ea5fe05b239e19b70 : SUCCESS in 5m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1e0b8e3918774d368a2bfa4da33653b9 : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b353f5f9101f4d07821623f2d9085a88 : SUCCESS in 6m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4815083356f94125b09b810c476569a6 : SUCCESS in 6m 28s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ce91aa5ed4c241b982b71513bb1a4483 : SUCCESS in 18m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8505f33595cf45c8a9b5be06b83149d5 : SUCCESS in 13m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/03e07dc1f4ca47a2920df025059a2500 : TIMED_OUT in 40m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ebfd1da30cfb4697b01ef1cbf5203019 : SUCCESS in 19m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/55bc1dc2872347a9b17ffc8c85f70469 : SUCCESS in 18m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7863931558374447a0fbb48ff7a3d243 : SUCCESS in 12m 47s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8fcb260f4fb749d7b89f114cfb7583ff : SUCCESS in 13m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4302f83a8ca54f468caa68eaf63c5a7d : SUCCESS in 19m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/399a5495ba204ca59cee0bdaceccb5a6 : SUCCESS in 5m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d14872c8042545ddb2c2a4af68c973cf : SUCCESS in 8m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ee6907dd6f143a5a9a68a6313d1f562 : SUCCESS in 9m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e92802d6d044648865f9d2e884ca962 : SUCCESS in 44m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/72424dc475924663a305dba31bc7d124 : SUCCESS in 1h 11m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4d1115b3b3c2423a92837a71f5844fe2 : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d7861b35cbce40bf9a9fcfdc80bbdb2a75c77d4e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-11 22:18:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dcf71c0e3f4cd4009d7aabb7b528544b26b90dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 00:01:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d067eefc72d4c799fc68f4b10a0b44f : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d4911174bdf34904a1f240dbe0e80282 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a257455ceb64c2faae0695a6ea958e0 : SUCCESS in 10m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da91c92bc81a4e97b7a85af22cb1066c : SUCCESS in 11m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbcfe7e8af9f4eadb1ccdf52be4e1bf2 : SUCCESS in 10m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21865b76f36b4e9aa2e171c81527ed3b : SUCCESS in 9m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6b5c29e815d94432ad7144d121a30da2 : SUCCESS in 30m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c776ce4cea7e45a2af65ff2a1f08b620 : SUCCESS in 37m 16s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c6b14fcf2ac1426796bd0b82fbeda04d : FAILURE in 31m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/368cfc7600a942ac8bcedfbbc7b27653 : FAILURE in 31m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/71553ff50b5247839973ddde870516d3 : SUCCESS in 12m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8cf9a1fb786743fc9d7ed453e375707f : SUCCESS in 12m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a3f8f81e171a41648de1a04207b0266d : SUCCESS in 19m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1bc0b06dd5a64f399273d5ebf0fe8c1a : SUCCESS in 14m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ac5836c90980439abf8681e70fa6c727 : SUCCESS in 24m 27s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/9f2ceec9307d440387166a77c0d966d3 : TIMED_OUT in 1h 01m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f1e5edadbb54980b4311a6bb46acdad : TIMED_OUT in 1h 01m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a46b1c2306a043249d7f2348029cdaac : TIMED_OUT in 41m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4f7a1d4826d14ec98955e4fdc03c9df5 : TIMED_OUT in 41m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3207674b889f4a47b6a7a7fc51171d5c : TIMED_OUT in 41m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b7e8d5e1113c40a3b422fcf3e0e92b9d : TIMED_OUT in 41m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e83d0ffefbbe402aa230f0dccef828c5 : TIMED_OUT in 41m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e3f934d806c4f1e8a8b8423c0e91855 : TIMED_OUT in 41m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d0799ae3046461fa1c25f0a1dffd177 : SUCCESS in 11m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f820c2ec2ed1420dae2e15426d9f994b : SUCCESS in 34m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bc4e7bb6d5fc4216ae744ba0d3d9c67d : SUCCESS in 13m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf03ef56638b46539b4789a19b53b0d0 : SUCCESS in 1h 37m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5c7192e508434f639ce9017fe5d0d1b0 : SUCCESS in 1h 41m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7429ed0d73ce40dca46466e5316b9f4c : SUCCESS in 1h 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"134502afea2733c8472d8ae4bbe844201d47d8fb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-12 00:12:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"03c995602f5b9df259e3081d8153a395b31cac99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 01:29:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/991083ff85134142bc131cf45a8196eb : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/267c6a9b62f042bb83f69f37341e5b90 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24b7f36335224128b3a9941922a9ed23 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac1858daa77c4c728be78a54effc822d : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/acbd189582cc4cca849b60e233b024b8 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b198a2c205945b5b231e0f6b3b0c280 : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/845860a62400470bae0cfe5378b116c8 : SUCCESS in 7m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/749b357052cc4753bd9faab7c1cdd83c : SUCCESS in 7m 03s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9f4673f7a72747af96ff32e8bcd719c9 : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/91a9b196254045b687c1d146edd9c68e : SUCCESS in 7m 56s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/663c034b22b34b1f8025e5fcc009c69a : SUCCESS in 4m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/54063eaa3c084a2d8a037b3fa51a6b28 : SUCCESS in 4m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/660c738712c34366b2216452d05e49b4 : SUCCESS in 10m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b5c603c532e541f29ff75983d93ece61 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dfed710bec144683903f567aeac14eb3 : SUCCESS in 6m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8691b09ac6564d2faa8284bcfcdfa12b : SUCCESS in 14m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/12b4a4a286ee4bd2967ae7d51e8bd926 : SUCCESS in 11m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8ef1afa2881244e39e86574a6a0e5861 : SUCCESS in 22m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/527b70ebbdf942d5be2a1d4082e4a677 : SUCCESS in 22m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2d9c13690f1549008614c970e8f8336b : SUCCESS in 13m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/822ccc4d07944d41a2cc592ef5b734d1 : SUCCESS in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7381e28b0c384b868c2464bcfc0919ac : SUCCESS in 12m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b29065f759dd4b5ea4dc1b004b3e2330 : FAILURE in 19m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87d86aab678847bbaa1a0b2a0cbe4519 : SUCCESS in 5m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/97212b8b18854ae2b5b6f8091671982b : SUCCESS in 8m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/46a1a0be8d9342aea901f78e37df2b53 : SUCCESS in 8m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4e38c1c140d4c578cd666edb3ff5e89 : SUCCESS in 1h 10m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7064ae930522435a8b18c172e71cbc03 : SUCCESS in 1h 12m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ce1f12cc9c20448d98da916ee789b6a2 : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fea69a0ed7f4b03c30a219e031a42167f6c65dba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-12 02:19:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b6b992dfbe0ccf16ff2b685cd547ec53fc674c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 03:44:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac49d4e8a6104e27bbf34099a370aa94 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cad81b00d3f4246abd31919a82d27b9 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3889b0455fc445ab3e6666ee46f01de : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73faa0843fcf4606b718b93c22ffa086 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1fb02600a464454cbeb082018478a441 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f38769d8a454a129541bcdc1c403558 : SUCCESS in 3m 58s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/5e555dc460c44be989851cc76ea81b15 : SUCCESS in 9m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8066ef8d477242ee9502485c1a6bdeec : SUCCESS in 9m 05s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fea434edec5345a085b49f339e500285 : SUCCESS in 10m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/97f43ce3dea64a3395ddad16d79f9429 : SUCCESS in 9m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cbcc89a8a4eb487da54c32c80c2071be : SUCCESS in 4m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7876ca691c93433f9d4bc66fc797b3ec : SUCCESS in 5m 54s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d16d3463764c48b190c755266fb57f95 : SUCCESS in 10m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f08661f89b20415faf7e3246334db28e : SUCCESS in 9m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/57dd7409c3f14770a38c3b8dc4472b02 : SUCCESS in 9m 17s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e6f5e6405b784daf9011202480b89105 : SUCCESS in 17m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8b2a49571f4449fcaa19ed62243db1a4 : SUCCESS in 18m 13s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/91ae807c366e43e6926a1c81a763a898 : SUCCESS in 22m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/abba20b90a8742daa8ca937d0daf7fbc : SUCCESS in 17m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e506f98626ec42a7961fed67aa9a528f : SUCCESS in 13m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e9125f09c6c9471c8b25ffb92cd191da : SUCCESS in 13m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/19fb8f357d744dc8b2a56426343da760 : SUCCESS in 13m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8baf25bbac4d432d85e56ea70feb28f3 : SUCCESS in 18m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/58d0300130a04f06a820918a75600658 : SUCCESS in 5m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/450517d0ca5946d6a1ae14bc3d8b487c : SUCCESS in 10m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/107c31d3d9404f38a4e6779071a050ca : SUCCESS in 10m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5d728be037e48bdbec03acdb29e7329 : SUCCESS in 1h 04m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/38fc1970fd4349f3820bcfb495f0f9fd : SUCCESS in 1h 22m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3fd8bcfcd16a403881080e0bde4acbd1 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e6662a958dafe6bda71801a8600cbd55ba0153a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 03:44:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef3a0a09f1ee7424905d7929e2c257237abb74ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 05:14:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef544d7e4644646855a5d1ae40f214f : SUCCESS in 6m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95d277bbc92b43f4bccb78061a684315 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8719fa3a7c8e47359337e1f3f04e4509 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/599217100a71431cbfe2e55097aa34d1 : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/56d18fb04eab422fa6e541c9da451c33 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f1b690875b64b918a590c0a82bf7451 : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/14cc6eda248e44d4be55715eba06b1f0 : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f540a6898c854804bcd035c8ff9c49c9 : SUCCESS in 6m 23s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7459abed47b44a6ea0f7ea53f1d979a0 : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/71e0dedd78f64455a120bce4078fdf67 : FAILURE in 7m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3729a44a96f94283ad318623cf550e4a : SUCCESS in 5m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/26c081bb4fde462c9e532c26e763cca6 : SUCCESS in 5m 06s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9ed2ab4168d14bd1845e204e7216aaea : SUCCESS in 9m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3568ea46bb3043bc8c9bc923e61f5397 : SUCCESS in 6m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c8db347658343a99e8c5057e9c0ac8f : SUCCESS in 7m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/71cb2813a7a0490daff03b37010770bd : SUCCESS in 12m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/028ee6bf36a443289f047c03704ad3e8 : SUCCESS in 18m 35s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d666b7f8a20b4f42a5b7f705c2cb4f72 : SUCCESS in 21m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e195d88c700f4680ad530f92dd3d7fa8 : SUCCESS in 20m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/67c90c7eb4a34c978a4a46382eaae92d : SUCCESS in 12m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f426bff6dc594b2d974d2a1753d36f56 : SUCCESS in 12m 49s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7d6d0c762a494e72a32db5c803db712f : SUCCESS in 14m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c949e395eb364b3bb8018250bbd19bd6 : SUCCESS in 18m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56130bd6c87d471cbe333708f542e677 : SUCCESS in 4m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9101ad5202e84a9bb7212febad3c0f25 : SUCCESS in 8m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3bf3b6bf36434460bd2f753a83946ac6 : SUCCESS in 13m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/66ec17a4fbaf4ac984a0f6f29c6629ea : SUCCESS in 57m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/60db7f2a48014ddf9b2f08f12763f722 : SUCCESS in 1h 25m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f0718ffe45ea4f85854bc2734e05be32 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c587962e781457a4247441bc14a5f65789f8f362","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-12 05:17:18.000000000","message":"Patch Set 1:\n\nrecheck\n\none of these times...","accounts_in_message":[],"_revision_number":1},{"id":"7a8c7c63134ae8822e1404fd243487af524196b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 06:56:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e4578eb76374ff1971d644ad80144a8 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad45a7f7662644229630b29d95542b0c : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/850a5eab8cb048dab274b2b27277e431 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d3ae5e36c9d4267ad2265d8b5419088 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4cc8c1c2da748a28b03118aa4cecf3d : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/088ab8cde09445ec837d88bf1f20132f : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8891c54344614be8b55ac93b0c258a93 : SUCCESS in 8m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6680a05bda9b41dc9f44ae0ce7cfb8d8 : SUCCESS in 7m 54s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/413f4eb0ee924fb99f44a9cb6e10ce03 : SUCCESS in 12m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f94432ecfc13471ab5e29dcb50472e3c : SUCCESS in 8m 44s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ac09ba72e6c4467dafe0b1c306db6a00 : SUCCESS in 4m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7d8dd26b92794ba5ad4626fd4849a552 : SUCCESS in 4m 57s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fe874f4c035c49368e55c42283343348 : SUCCESS in 10m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c6a930e3b630459fb40aa6e9da579d1c : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d87a6afc69942f9a672b92a878b8738 : SUCCESS in 6m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/033f0564d3fa49cd8c4c477a0f0f0b76 : SUCCESS in 13m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/af53eea8a2924bfe936f612b3e876fd5 : SUCCESS in 16m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/91393c3096a5412ba267c200aebbab7e : SUCCESS in 23m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/33d54a40bb0d44c088daffc6fb45a2ca : SUCCESS in 20m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/77d00cd030904697a2e59a1c3528949d : SUCCESS in 12m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/174d7b24b61c41df8aed7c1a41c28b76 : POST_FAILURE in 16m 32s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/23e95fb94af948a08303b7d325de6f8c : SUCCESS in 13m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e87bdbb332bd4bbf8bd2024de3a606d4 : SUCCESS in 22m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/583a96cc56204e8bbfa09d4611ad1ddf : SUCCESS in 5m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/feb35b94a64f42f2b3dff4719025ca95 : SUCCESS in 12m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ed8aad491ee41d7a605edfcbc660725 : SUCCESS in 9m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/94985289400e4a968a83b1dec6e2eef3 : SUCCESS in 1h 18m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ddb5d0af493c466dab8dcf1d4a84b206 : SUCCESS in 1h 34m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7560f30dcf934a0e890464b58430b4f4 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6872489369f3504dc21819053db14ed0516cb0d8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-12 07:01:00.000000000","message":"Patch Set 1:\n\nrecheck\n\nnot this time","accounts_in_message":[],"_revision_number":1},{"id":"b8e604861bfab6076ccee2b381b6488c5e6b4ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 08:43:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b6f38180d249c8a5533bbeca72245e : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/275c73e93eb244aa84e549ca1a6d4d40 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07a4303c93a445c39cee818178981464 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab80fff965e94d03bb919554cb5e69fd : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/359c0efe9de54540b8ed4ba77b0255d1 : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/deb9c0f001d5417bb0866aa58709eead : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/35f73f9aa29840b28d707c2f63528d08 : SUCCESS in 10m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb9f8162168c4716bb83212831e4acba : SUCCESS in 9m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/958e6c82637e4354a74eafe1c2456907 : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cf375459af1a46608f2510589b5dd82a : SUCCESS in 13m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/36d0955920ed4d3b82ac22315322c22f : SUCCESS in 5m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7ae2422d3d5549fdb8bc2f5aa1aee8f9 : SUCCESS in 5m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a518c6ad21c84276af4c507418926877 : SUCCESS in 10m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/726171a94a15488abfccc5caf8ae3690 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f1aa2918a5b64cb9a104be0930b4915c : SUCCESS in 9m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4c4034c5e83a43a7a241b774ff9cddc6 : SUCCESS in 34m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f1cfc8a8ca9d46c695bf60e4d884ebb7 : SUCCESS in 13m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/32fa85683e994a659656c0f7583d3890 : SUCCESS in 23m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bdd0afd1040d493d88af31a4ce55dbef : SUCCESS in 22m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/17fc1553b4cb4d2f97b6c22bd6b7ca3c : SUCCESS in 24m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0b9ffa0b03cc447eb2a46b2e50918067 : SUCCESS in 12m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/25b802842d574540901f42919ed04062 : SUCCESS in 24m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/61c796e814264d95a08fbe9cde57218b : SUCCESS in 19m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72c9d46f3f6449d4a1834e0d3bd2c173 : SUCCESS in 7m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6d604e294f67406a93da9ee9328cb320 : SUCCESS in 11m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5af89184b3204440bd43eb27561152db : SUCCESS in 17m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f33779c63754b3c83ba3a719bd36495 : SUCCESS in 1h 07m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/02a7b9a0502b49e7b087340066472c90 : SUCCESS in 1h 36m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4e9e83a44dfa40f4b2fee1130190cc25 : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8fc386cabc5a902f0c2bb74c7dca95f56753b86d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 08:43:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"383d2873f2b849167d961b02b91f0959f451ccc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 10:23:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3c7ec28a24412c898245ba6b9bf359 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26215a92e1d14eba870d244ae2c0a426 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a31faaa0c58241bfac04e8d1180db0cb : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b0279ca26ac43099b72edc8d38d64de : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50367b4bc00041989e5811f4e004d86c : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2ca4027311245c29b458663f6d14649 : SUCCESS in 3m 36s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d0b1ea3a1c7644f8a2693e210ce7f73a : SUCCESS in 8m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c6481ae927a4723a4c5612538c6e44b : SUCCESS in 7m 31s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ac3bfcb4b5644aa9b98df709a57fa684 : SUCCESS in 9m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/702506c4177747d0b00881ddb26f9a6f : SUCCESS in 8m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/71577eda56114be5a2bd9bbe2c2886ed : SUCCESS in 4m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/22950f504c7f4933a6c4da0bd35cf9d6 : SUCCESS in 4m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/228e748b60a542fdae705d806ea45903 : SUCCESS in 10m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f14f07b7c08144afaef4c98426b9d41f : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f4937afadbe4858b30bea2e3b76ede2 : SUCCESS in 6m 43s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6a8d0511457c444691801390352ab0c3 : SUCCESS in 13m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ac3b75624bfc4c6cb77a240ebacbbb6b : SUCCESS in 13m 27s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7c6da3dedec54ae69a4af0cb66444b0b : SUCCESS in 20m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0f50ed806efb4b4a91b8dd7ee38a9367 : SUCCESS in 21m 01s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/42420cd07b1a47d4896057fe7510cb3f : SUCCESS in 13m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/20101d63eeb74926bc5f87b489da181f : SUCCESS in 13m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/64a80e11feb64999b0f6fc2a092bd297 : SUCCESS in 12m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9a7f1ebaa0964b2ab1db9f57a5e604c2 : SUCCESS in 20m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c02e4db552eb4c479fc12410d9208d77 : SUCCESS in 5m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/680531c41028419f92837bef9141fa72 : SUCCESS in 8m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/495049f0642343019cc0472785c99dbb : SUCCESS in 9m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/004e5813053249b0bdea53f6c3fbe4ee : SUCCESS in 1h 04m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/63b0a8a32ba0410997d40d6ebf940272 : SUCCESS in 1h 34m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/108ef264549b4754b26d40e1db126457 : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8c406d30ba3796e6e95dcaa9e51777e74eddf8e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 10:23:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33dac5534731434f04ebe0a9482a315c7d24dc4d","revisions":{"33dac5534731434f04ebe0a9482a315c7d24dc4d":{"kind":"REWORK","_number":1,"created":"2020-07-08 18:57:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/740089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/740089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/740089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/740089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/740089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/740089/1"}}},"commit":{"parents":[{"commit":"77ac7b164f7d294ba80832324f4463101cf3fad9","subject":"update constraint for oslo.utils to new release 3.41.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77ac7b164f7d294ba80832324f4463101cf3fad9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-08 18:57:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-08 18:57:01.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 2.3.4","message":"update constraint for oslo.policy to new release 2.3.4\n\nChange-Id: Ie39ebcb919e0b56ee7ea04f97a4f317ff010d030\nmeta:version: 2.3.4\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I72c703037acfd2eb591d70426be06a3fff72123e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Vishakha Agarwal \u003cagarwalvishakha18@gmail.com\u003e\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33dac5534731434f04ebe0a9482a315c7d24dc4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33dac5534731434f04ebe0a9482a315c7d24dc4d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
