)]}'
{"id":"openstack%2Frequirements~740097","triplet_id":"openstack%2Frequirements~master~Idcf20248c528774bf59aa9a0c2e39bd2e339a055","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idcf20248c528774bf59aa9a0c2e39bd2e339a055","subject":"update constraint for python-openstackclient to new release 5.3.0","status":"MERGED","created":"2020-07-08 19:23:22.000000000","updated":"2020-07-10 03:09:58.000000000","submitted":"2020-07-10 03:08:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740097-1594350517783-64565078","meta_rev_id":"fcd35172106330bde4c7d04df09439d69faabf73","_number":740097,"virtual_id_number":740097,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 03:08:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-08 19:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-10 01:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-10 01:23:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 19:28:59.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 01:23:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-10 03:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b690f6e06ca6b5a8e536b5083ab2fff24d03cea","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-08 19:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"037acaa052019c5411a60c81ee32d633d1d2a419","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-08 19:28:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c20ee8863c57abe0790ef88a67149d3aa1f0258","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 20:09:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d7477d710b5448c8bbeba279e680ce60 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":1},{"id":"71abce3d5238b696ef2122dea45ee2bb133d221d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 00:05:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c06cf72b9b2f42d088b9899e022e5f24 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61b5420f7254eb4b025d49ad1d07db7 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94fd9299a8fd4951b5d9708dc3321bc8 : FAILURE in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/934304f13de640fc86299e20395ac7cd : FAILURE in 8m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04f950ab39844880abb337d9b16d8097 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/285bf3f815d941c4b282d30294fa6d18 : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f80f437aafe4db5adc7b66e29940646 : SUCCESS in 8m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/86336f1ea06d467fbb7c3cacde9a31de : SUCCESS in 14m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b8e70a2cd8144db2baea5c6a0d866fe7 : SUCCESS in 11m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/719913094b7649b282ad64c8cbb7ab9c : SUCCESS in 6m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ed28a4c9ec494a7fb13ab7e332a25e61 : SUCCESS in 12m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ab941a72a5fb424a8f643ecf0e469047 : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f32fe007f8af46dbaa0a861cb93da756 : SUCCESS in 11m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eb58ea78700841099eb695d2d60b440e : SUCCESS in 23m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/703829cd0d9d46c9b411b028ea64e203 : SUCCESS in 27m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a884704c12fa42d69dd76078bf665d32 : SUCCESS in 15m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7dd26e1de4744278a9fe715ef9b2832 : SUCCESS in 17m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1d28948c18fa44e290750eb119b17d11 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a959573ac764def81b935f53373f071 : SUCCESS in 7m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/79d13383b8d74bd9891ee98e740d4d0c : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb02d0e4b25d48dbbd05c5146b9079da : SUCCESS in 50m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b698fa4e5524e7ab0d0e14aa92f3ac9 : SUCCESS in 1h 25m 18s","accounts_in_message":[],"_revision_number":1},{"id":"03395195c8f29c93db4396b71995eb6e4e905058","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-09 23:09:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"748f1b5df39ff026619e2346e12b3443f601a820","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 23:59:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1778f9473dc4410d9bd9296716ca6532 : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e62bd746a60b26eab3057c328115db67f8a388e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 00:41:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a44159e78db4d24bc05829b69a3ff61 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc4e3e5bf1846938f61da9f8858d2d6 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea52ededfb4341bf94706a398281f962 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e66e5e9ec1f34491ac51d8237c4b1adb : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83e5498fbf8142dca342fedec431ffdb : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/595a0712676b4e56959f5fe4768f680e : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8bca97fe125143638c705eeeabd118d4 : SUCCESS in 8m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0f5fbc5e501f48a5bc2e4b5551369721 : SUCCESS in 10m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8803923f03834bd5bc25d332bd1cf139 : SUCCESS in 8m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c685b50071d441dfbdb46d201280492b : SUCCESS in 4m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/157039670f5941788b0ba22c0a59c764 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da032ac9feab4ebdb836db0b989f4be1 : SUCCESS in 9m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f979490fc02e4e15920e33f3f9c97b45 : SUCCESS in 7m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/91ea9a6f8d674a759fc6aa70283c84b4 : SUCCESS in 25m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fd58e306ebeb40f897f448485aa3c254 : SUCCESS in 22m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d76ad8d1d5b04c21915e7fad28a2dd36 : SUCCESS in 14m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1c5c857ac3244cc7bd2e7a6758fffd4e : SUCCESS in 19m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cab07500f486423c896d4a17a03c6f39 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ae046395e2245fe9a760cdea5b271be : SUCCESS in 8m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dfe0d92a513c45288c41d064782dff61 : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ccf77ea05dc4f219846c05e848e2116 : SUCCESS in 1h 02m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d36001950aa425d9ed35df70c8b7b54 : SUCCESS in 1h 30m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b8c363863ccf1932bbe7fcf3bdd688bc99cf83f2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-10 01:23:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e0e57bbff45b9bf2f6986259956866cb0131475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 01:23:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c562c4de9ea5cccc500953188c5ea40fadb78f7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 03:08:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4e097560a8d5c15096c044b692fd8f3f930bfcae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 03:08:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b788edcc6a114bf2b500eb4173b5290a : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/245fa4a1b8f844d89a5da3ca2f9a573d : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41eb872a8f5d4bb6897b9c0d4e84d77c : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e7ed0fa7512446b899dfb7e4c44da12 : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/830f6b55917845fab8bf6d4406ae51b7 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/03e57c4ff0664fe7b5f0ddfd08684e96 : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ba3dd1d8a5664d18a88c8348a2af5432 : SUCCESS in 6m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c18add9aa42d40b7bff5f70c312906aa : SUCCESS in 11m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d32932aa8cde47eeac4798ac741cc199 : SUCCESS in 8m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/05ae8da093634f1ab647b60f762bf9bd : SUCCESS in 4m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5fd2c759a22f46db9cdb88027b106264 : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e50ddf3dd89491b9571e69194f2ebbd : SUCCESS in 7m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/97f13bbd5e2d45118b1c19b6b58809c9 : SUCCESS in 7m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cbaf8dbc8ea14941b6e10b7f30515d99 : SUCCESS in 17m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/afd25b38d9494436a28e9fbd07451ab7 : SUCCESS in 27m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8be8013215eb43db9c77fed801bb3861 : SUCCESS in 12m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc7d0fba01df4c6bb1606560d0d93abb : SUCCESS in 16m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d4a97269f1044fa28adff661633340c2 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ccafb4fe2750406e8d1ca4685f78962f : SUCCESS in 7m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2d0a0f2189bb4639a5740fdab53960f8 : SUCCESS in 9m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7a90eff6fa944ddf857ecbae6146f592 : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ed2801ae6fb48229da6cb5c40908b82 : SUCCESS in 54m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8fdbb0a7f7e40789a4d3cb7371d1035 : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":1},{"id":"99ce62317bb9b2d56cac3c8aa1e0c68f77ab0ba1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 03:09:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce35e2c1139f4f9bb54ec468e1f8b9cc : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f74bd40fa5dd1be47eff8bb3d5e253b03dafc3e3","revisions":{"f74bd40fa5dd1be47eff8bb3d5e253b03dafc3e3":{"kind":"REWORK","_number":1,"created":"2020-07-08 19:23:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/740097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/740097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/740097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/740097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/740097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/740097/1"}}},"commit":{"parents":[{"commit":"017cb7c83f9c152b66011a035441a5b1e7998177","subject":"Merge \"Add \"isort\" library to blacklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/017cb7c83f9c152b66011a035441a5b1e7998177"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-08 19:23:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-08 19:23:20.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 5.3.0","message":"update constraint for python-openstackclient to new release 5.3.0\n\nChange-Id: Idcf20248c528774bf59aa9a0c2e39bd2e339a055\nmeta:version: 5.3.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: Ie4adba1ca399abb43abc7741e55e82968ba742f6\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74bd40fa5dd1be47eff8bb3d5e253b03dafc3e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74bd40fa5dd1be47eff8bb3d5e253b03dafc3e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
