)]}'
{"id":"openstack%2Frequirements~740198","triplet_id":"openstack%2Frequirements~master~Idfae539888ba94549617811ae4bcc1c7ef0ccf87","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idfae539888ba94549617811ae4bcc1c7ef0ccf87","subject":"update constraint for castellan to new release 3.3.0","status":"MERGED","created":"2020-07-09 09:16:29.000000000","updated":"2020-07-10 02:24:19.000000000","submitted":"2020-07-10 02:22:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740198-1594347764829-c502fe0b","meta_rev_id":"929fca577de434390d547463ee70279fb0a6bcbc","_number":740198,"virtual_id_number":740198,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 02:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-10 00:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-07-10 00:31:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 00:31:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-10 02:22:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0d26d8b0bf3c12dedd6e898dc95600dc05d7334","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-09 09:16:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"275ffadd2e65ee1bbade85f6cfe5586cde37ca7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:59:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3b88f260788e46e887be8658124e567e : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8964bdb630225c5abea8103f663b2758c64da3b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 11:30:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/391b1e86a32141debdc6a003bbeba5d6 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/790cad4b749d41409ee622e11e60f223 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b1159196b0b4b91af987ceece5f22a3 : FAILURE in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/182607565cbb40e9976c2850e1210122 : FAILURE in 11m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3c09b3420e646d3bce96d786e115a80 : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d21c6de2fe344ee972939c208eddead : SUCCESS in 8m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7671304f61f04302847a175b2f0474f2 : SUCCESS in 9m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da67754a2e8e4984bc282de67b634976 : SUCCESS in 15m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e759fc69a18f4525b6751dfb41e5cbab : SUCCESS in 12m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7eb75075efb14df88212c1435ede84c5 : SUCCESS in 6m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4a071f5bf7094bcaa6422d5994acb80e : SUCCESS in 17m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ef02edfdd354aa19c4f443c8f80ed33 : SUCCESS in 10m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3a04feddd0804c0db88407ecb995ce3b : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7b25ac27ad154d64808fe91265afbbd8 : SUCCESS in 23m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/32e01da2dfe0455aaad5c8736d6afe5c : SUCCESS in 27m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b74ac297c11c483183a7d85402a83637 : SUCCESS in 21m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/69d2191987714b10bba5805a5f91d978 : SUCCESS in 18m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5197c10b2eb64d54883a10ad57c05267 : SUCCESS in 10m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/151a027203e6458f9ab93548578906aa : SUCCESS in 12m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/60108a5af8b44e439610e77fe6e5dde4 : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74767ef417e84623a5feb0a3f4fb05e7 : SUCCESS in 57m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0dafdc3d7f314727bdb99d6257995614 : SUCCESS in 1h 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"659b66c2071628273d1ce6467da669f0e69932e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-09 23:09:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b442720c19fa67ddcb62c68cafb76b52fa661461","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 23:52:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1805862e47c94812a4bb1c28285412b4 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":1},{"id":"56f9993482281ff552f6c3989c1b312ff946654d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 00:30:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3546dbf3284e4de1a57aca20fbe273af : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1252fb73247f425dae059ab693dc195e : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a788c4b8869425ba211a45315022f54 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e68d3e9e09f540d7a026d3700959ec0e : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d4da528212f4cf4a8638ea87f6b56be : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce36d795dd9946f9b415d828aae90161 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/33c60d8ac39a45169fe2b70e4260731d : SUCCESS in 11m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/abd1273b3a834d10a5d001edb473ee9a : SUCCESS in 9m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3986bf8a47cd41e1b8dd0be3dd250a43 : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/29d901986a284fa2807bc3f04636f881 : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/107a7ecb761f49df82cdfb738f4010df : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81515e4cf3ae4dba9e06bd411db49e18 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/252049d953a648799f213e1c989e2ff6 : SUCCESS in 15m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0b9abca7ec504f55bb34e5996c4edb94 : SUCCESS in 13m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2ed3394dab474db69949c77854153dce : SUCCESS in 25m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/72a30c419f0f4fffbe4a9ad82fe9b0df : SUCCESS in 12m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/802d007f17394e4aa34a6cc357efeab7 : SUCCESS in 18m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4d838c0e6048416fb4caa18f03b2c2db : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3149219950d648cfbe1f70824a0a475e : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee27c8e4b0574d3dad28912f2f6218c3 : SUCCESS in 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe8f6fe854784e03aa721f98dc4fe4aa : SUCCESS in 56m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfac07fee07d47208c26b7ad2909ccc4 : SUCCESS in 1h 17m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9f127f1fd0c998285e0ae14ed136d9e640e4302a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-10 00:31:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1decced454dcf359277c2aa3f5a5c6ddb80d9c8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 00:32:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c1ce0316220bcff66b16f297f42175a7005cf46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 02:22:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14ffcf4e6a9d4c4b89bcb5b5bb610d19 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1940953bdd5c4aa89e880d6ff9fe0c6d : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0733d45098de4d8db79479fa39277979 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a0919d93e6449c6a0111ac82cb132e7 : SUCCESS in 8m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4a50a8c2d2c471286ea938a0e136399 : SUCCESS in 8m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fdd311f964c442f38e99b94c543d1f53 : SUCCESS in 7m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/713601a21e584cd195d1bc0652b7adf4 : SUCCESS in 16m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3e65eb67f9cd4604a878c5248413508b : SUCCESS in 18m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/08c646d4b5ed482eaba8e8d1286bccdf : SUCCESS in 12m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9ee8a22258ed410f8e349ce10c9511d1 : SUCCESS in 8m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/10c9d045edb04ba9be60a3187d967a89 : SUCCESS in 12m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c4cbf65e181410b99f3c475779fdaad : SUCCESS in 10m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a4d3bb70d0b441b980cf252dc432110b : SUCCESS in 20m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/297874fa997841aa98dc1c9a501e6e8e : SUCCESS in 1h 00m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a2857527fe5f4ab49aa2554399d84868 : SUCCESS in 31m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7cdc0af3a6ff41bca0dc05c200c63ae1 : SUCCESS in 26m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2666fe853d924cfd844383ba9906ea98 : SUCCESS in 22m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2a441ce88a1541bd90ffef6faae81b71 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/96d83ac65f754f6d88f2c37571fb1b42 : SUCCESS in 12m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ed4c9fa1faec4915a87e7609f65e4459 : SUCCESS in 12m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/818409f23ff346648652c6a8b6d446e1 : SUCCESS in 22m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8fa90e8822b4c13bae7e55c00779d7c : SUCCESS in 1h 22m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0702b08bec7440778dfa158edc492f4f : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b18194415a706510402a7e0e7fc5f6cf10547c50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 02:22:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ebc5de69df3731a9992aa13b1c3305b6e9f37cc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 02:24:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344bf900b7f04265a21254650e644dda : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a18c3661da352fc577513c89e0be87b9acaaebc7","revisions":{"a18c3661da352fc577513c89e0be87b9acaaebc7":{"kind":"REWORK","_number":1,"created":"2020-07-09 09:16:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/740198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/740198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/740198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/740198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/740198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/740198/1"}}},"commit":{"parents":[{"commit":"017cb7c83f9c152b66011a035441a5b1e7998177","subject":"Merge \"Add \"isort\" library to blacklist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/017cb7c83f9c152b66011a035441a5b1e7998177"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-09 09:16:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-09 09:16:26.000000000","tz":0},"subject":"update constraint for castellan to new release 3.3.0","message":"update constraint for castellan to new release 3.3.0\n\nChange-Id: Idfae539888ba94549617811ae4bcc1c7ef0ccf87\nmeta:version: 3.3.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I23bd10c2b1df4cf8a0cb6726610d5bd40e391a24\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a18c3661da352fc577513c89e0be87b9acaaebc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a18c3661da352fc577513c89e0be87b9acaaebc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
