)]}'
{"id":"openstack%2Frequirements~740203","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I3cdafad92ffef06c5488b3816f367b6919e4dbac","project":"openstack/requirements","branch":"stable/ussuri","topic":"bump-magnumclient","hashtags":[],"change_id":"I3cdafad92ffef06c5488b3816f367b6919e4dbac","subject":"Update magnumclient to 3.0.1","status":"MERGED","created":"2020-07-09 09:24:43.000000000","updated":"2020-07-10 22:14:54.000000000","submitted":"2020-07-10 22:14:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740203-1594419294608-9763ba5f","meta_rev_id":"6c164f91939acc342f2c89507d3a1e4625103d4b","_number":740203,"virtual_id_number":740203,"owner":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 22:14:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-10 17:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-10 17:44:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-07-10 17:44:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 17:33:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-10 17:44:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 22:14:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54779a8d1897a9496f26b6979ee25981084a2fcc","author":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"date":"2020-07-09 09:24:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01bfaa682b7588d7f0f36223228fc16977283c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 12:14:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d7d49ac2134560960d2342ad00d476 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/456171f1bd5f499e99d8ef0ff3bafc44 : FAILURE in 9m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/efbc749dd9a2409a8a7856928b962424 : FAILURE in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7e9cfb257534371922a247bdec1d06b : SUCCESS in 7m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d7df747dd9643afb6ed32548b1fac38 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3d1c4186bca74500a132c7a57ad1de79 : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bbc1430a3d914cb48a3f836777992527 : SUCCESS in 13m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dc88f3cfca2549bea4fb42a4fba5ee2d : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b41ebfacbd30436dab28a0bf01234f28 : SUCCESS in 13m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf5ccdc9cd25491aac3f2efa04e0a594 : SUCCESS in 8m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/345891331d3a4f618392e154c755d22f : SUCCESS in 26m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/983c38bcbe8b47bcbd513bb3735cd822 : SUCCESS in 30m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a81ea2546ee940eb8d736d87a0e56986 : SUCCESS in 15m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cbd930eee0f44a90884b48408bd3f85d : SUCCESS in 13m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd9b91f6b1d741f58ff6c81e51019482 : SUCCESS in 22m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/960619539fa947918c70a7ade3e57a27 : SUCCESS in 10m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/4e276ebd75c2478d9cb797235e5c39aa : SUCCESS in 10m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a01ace531f324fb4a500b992d1b20088 : SUCCESS in 11m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/81a5eac05d1f46fdb2084224fb0c245e : SUCCESS in 17m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4d59137859854c53af68da504456eeb2 : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/daad4b7be3f24008baeb474dc5b947ec : SUCCESS in 56m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef9385a52c83474f8be578ef856a9f1f : SUCCESS in 1h 25m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4adc0baae864f59a7aa8f4cdf8ea6eb : SUCCESS in 1h 17m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/9f2d0c129f3440ac9ba998cbdc6724c1 : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"564226508a3a8a208ded9420b036ec0e683fe67e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-09 12:44:45.000000000","message":"Patch Set 1:\n\nLooks like the update was missed when 3.0.1 was released a few weeks ago and there were issues with the job automation.\n\nSomething going on (pgk_resources issue?), but once we get that figured out, this should be good to go.","accounts_in_message":[],"_revision_number":1},{"id":"56dd3b1017056cd781d493c58c539b932c11471b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-10 01:23:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c723a486b4911cbcfd3accef342dd48ef2cb5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 03:29:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fab6c667862454fb900398234f6e5fc : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/243f844ccbbd4b0c9cb1656bdf9fe755 : FAILURE in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84dbfd44c85a4cfdbfc2fbd5a6883c37 : FAILURE in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/528cb4d84a32427eb9c5b94eb76bec25 : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8331210101b94a978ac0dc32a1ef7a63 : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2789dac3ada843dfb37869ae4552a938 : SUCCESS in 10m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/82bfb862a9f0459bacd2bca8dc375368 : SUCCESS in 8m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/958ee8af275241b295e8a02a3c8bda42 : SUCCESS in 6m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/27a5538f3a434852bd4c6daac291f320 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86a8768cfd9c43c882b279ad710af935 : SUCCESS in 9m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4f41713de0b145fd80f0c64d5c49ee81 : SUCCESS in 16m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bff54403964b4c3eb69b60d8b927455b : SUCCESS in 16m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/249b222789ef4ad58fd4b48ef547d6aa : FAILURE in 28m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/df4770e269a34611bd10d62714e22095 : SUCCESS in 14m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94f6c1033f714c3b874674038b716200 : SUCCESS in 19m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f3ee16d85d5646f7b5882ef61ca715e9 : SUCCESS in 5m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/41fceadd2e7d4021a1df2ff7606d0b60 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6bfae6ad75a4f869b5984f11d8beb34 : SUCCESS in 8m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c72787f3fe6b4b3d842adb7d736b9589 : SUCCESS in 9m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dc7332f276574e2da91e02a4d53abf89 : SUCCESS in 10m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/defeb86a632b4eaf82948cc8669368db : SUCCESS in 2h 04m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7108f720448e4517931bfcbc86f11a53 : SUCCESS in 1h 20m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6cec652af55544eda2ed82f540768fc9 : SUCCESS in 1h 20m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/7696bd9f2573429d8fabe32d6696d84e : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a3eddef49f3064cad02e8d327903a739d72281a0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-10 14:46:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de897e1889b7eb9e210d90824cc662e84135aed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 16:44:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dd71d0e8b9c4ba1b6377d4c6bf493da : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bf0c2807bd8466fa6c58cb838d0ef81 : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/487279875d85469eab32b64e22f742ef : SUCCESS in 7m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93ccee76399e43c1a6be9edeeb74c129 : SUCCESS in 6m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6dcee72e73da427fb5c5289dc23ec604 : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c83ea5cedba447ab77a4085eee12166 : SUCCESS in 11m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/07e89cc7c9274ac59cd16e24bc8fd520 : SUCCESS in 12m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/75e763b5e3ee4321bf66a32b5226c404 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d7755cd0fa564606b6db5ae38f5b9bd4 : SUCCESS in 12m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/36357889be6a401cae7a58b70d3b8609 : SUCCESS in 8m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dc74694b34fd436fab2d6f5cb5281b99 : SUCCESS in 15m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bfda6cd58e9d4abcbd0173014e1f763c : SUCCESS in 39m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6d5fbdb6baa84a538a89b99d0ef37aa3 : SUCCESS in 15m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/71a4dbd4d13745f9b36f0385ab77639e : SUCCESS in 18m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca88fd00e15748cea4c109143e1268fe : SUCCESS in 17m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b8581e8cfb9a4f63ac5bd7c50fd5253d : SUCCESS in 6m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/95e276149a2a43fcab3745f4130c328f : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c3aed51b28046bfb102bcd10e1abb8e : SUCCESS in 15m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1c917caa51af4dbdb7dc23dcf4a002c8 : SUCCESS in 13m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/978e326318804bdca9eb6d51d5f73ddb : SUCCESS in 14m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c3c4af8ee164e50b6fe6628539846f0 : SUCCESS in 59m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/74d91bfe2a794f1281e6dde85c081c4f : SUCCESS in 1h 29m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a63460cf23d4f3292982852d7922913 : SUCCESS in 1h 38m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/62d1ae02f05542fc98eaec2bf9ebba7f : SUCCESS in 1h 28m 18s","accounts_in_message":[],"_revision_number":1},{"id":"516433780f4fb6c6770afc7d47fc6858c5b7d549","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-10 17:33:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a1476554074a9455de4f34fe7032a8b719cd76f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 17:44:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d3018e996e11696b451c7aadb9443867fe3dbaaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 17:44:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"028966c3895efcc5363c5023738082a7aaedb0b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 22:14:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"77571f5d40a31925164d11fa27bee9ee3a0ab801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 22:14:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200ae6adfb9146e4818c707e8218cb58 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86cecd56eee149599b3fb9d392ddf6cc : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54d9f39e6f294b538a0375ca7e868322 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b15bfbc853fd4b30be414d42eddcd488 : SUCCESS in 6m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d2d94e4d6fc48119dcf6c3a68cb6f10 : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fa47bd68e0784229b4a7638c5ff8c05a : SUCCESS in 7m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/80e33d54cf01416497db29e1ab4c8cbd : SUCCESS in 10m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/29ebae0475ea44b98d6819a4b15225b0 : SUCCESS in 9m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/49ecf82363444574baacfe223488689b : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/55bd8fb4b0034fb882cc46bd670e07f0 : SUCCESS in 7m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9041bad9dd0646ae9ff34fa858698d35 : SUCCESS in 12m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/af5e159c30854d12824131eb00a6e7ba : SUCCESS in 26m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3ded1588008f4974a6b58e0510c9d0ec : SUCCESS in 12m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ead20ccf981a4bbf9348da57c3c40323 : SUCCESS in 13m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b29ddfbaad046afad789966e313b959 : SUCCESS in 16m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7696c3abed40420394fb8f8ce7c62777 : SUCCESS in 5m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5fddb57e5e3740deb0776039fa93319b : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/26c534a86bf24229885c4d78ebe60581 : SUCCESS in 6m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3965e6d248dd4735bc5d5fa646be5caf : SUCCESS in 9m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/04e0d622a12c4227b20cd161dd2ffe75 : SUCCESS in 16m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d99cfa7649aa47e8a34d71ef57f23095 : SUCCESS in 54m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7c62e0206ad4a38b5be2144b5d6a6c1 : SUCCESS in 1h 18m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d07ad04a00b54250a55c87aa5154f55c : SUCCESS in 1h 25m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/3fa629870ba2472b8a4f59db344959ae : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3c9ac764649285a8455d9cfebb0f4cdafb1447f","revisions":{"e3c9ac764649285a8455d9cfebb0f4cdafb1447f":{"kind":"REWORK","_number":1,"created":"2020-07-09 09:24:43.000000000","uploader":{"_account_id":20498,"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","username":"strigazi"},"ref":"refs/changes/03/740203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/740203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/740203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/740203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/740203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/740203/1"}}},"commit":{"parents":[{"commit":"42daf674222e85fb4dc3d4d1bb113389ca3a7137","subject":"Merge \"update constraint for oslo.policy to new release 3.1.1\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42daf674222e85fb4dc3d4d1bb113389ca3a7137"}]}],"author":{"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","date":"2020-07-09 09:23:38.000000000","tz":180},"committer":{"name":"Spyros Trigazis","email":"spyridon.trigazis@cern.ch","date":"2020-07-09 09:23:38.000000000","tz":180},"subject":"Update magnumclient to 3.0.1","message":"Update magnumclient to 3.0.1\n\nA *key* feature (merge-labels) that is realeased\nserver-side is missing.\n\nsee https://review.opendev.org/#/c/738426/\n\nChange-Id: I3cdafad92ffef06c5488b3816f367b6919e4dbac\nSigned-off-by: Spyros Trigazis \u003cspyridon.trigazis@cern.ch\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3c9ac764649285a8455d9cfebb0f4cdafb1447f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3c9ac764649285a8455d9cfebb0f4cdafb1447f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
