)]}'
{"id":"openstack%2Frequirements~740424","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I5bb01fb21982fb8757e925659ad206a15a79f799","project":"openstack/requirements","branch":"stable/train","topic":"fix-tox-stable/train","hashtags":[],"change_id":"I5bb01fb21982fb8757e925659ad206a15a79f799","subject":"Fix failure with setuptools \u003e\u003d 49.0.0","status":"MERGED","created":"2020-07-10 09:06:54.000000000","updated":"2020-07-11 06:14:27.000000000","submitted":"2020-07-11 06:14:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740424-1594448067335-2400accf","meta_rev_id":"4a88084038723b8f880514a0147826a287129912","_number":740424,"virtual_id_number":740424,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-11 03:26:40.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-11 06:14:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-10 11:22:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2020-07-10 09:13:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-07-10 09:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-10 11:22:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-07-10 09:06:54.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 09:06:54.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-07-10 09:13:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-07-10 09:52:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-10 11:22:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-11 03:26:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-11 06:14:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"573ff7da67a71fdea07c872418e1a25fc3904b16","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-10 09:06:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"bbd7114776f0e16658acbd0dc0d08f3f9972aad2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-07-10 09:13:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"97d5de2df6ea864eeb0193f2083dd9069fb36513","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-10 09:52:34.000000000","message":"Patch Set 1: Code-Review+2\n\nClean backport","accounts_in_message":[],"_revision_number":1},{"id":"5c36ab64562ff13a0bfc228db1f23746b962cfc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 11:02:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9870129e27ff4ba38fee78fe01f45c37 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acf9ef5f08124066a080f11cd4f73afc : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23ca1ee9baf94803aa9237de21e26c58 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0bb9a50810448bea0c450fd4be9182a : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ad512d6166044b6bc91097b47276678 : SUCCESS in 5m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdd3c3905d5a4663826e9a6070927570 : SUCCESS in 1h 06m 23s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d77fd20c8f2a4225b2e3d7eb46e51d17 : SUCCESS in 1h 37m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6fc3cb1bfb054bb194989299e0c99fa5 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8bacd60eed1562608bd36ea65aac9440a197413a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 11:22:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48b01aaa21aa5b8035628382ad57de7e9852ba03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 11:22:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ea315fac55642958a30c3a2747e85cd1b762698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 13:23:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c46f778c614846468585ec630543ffa3 : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9eccdcdb793c41dd9644d411c2b03304 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/744231686219408aae1779061947dc9a : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df772fe4f7364b10af418b24f13fe2eb : SUCCESS in 6m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36b55386c7d448f29e32527a7613219c : SUCCESS in 10m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b8a2efcedc24686816610daa09354e5 : SUCCESS in 1h 03m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c0c14d4e5c294598962145b657f65068 : SUCCESS in 1h 04m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d040c81db6a341e38b1c221a7a83b25d : FAILURE in 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"736a515cede190c1111ac644c594a54005cd81f6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 13:37:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d93bfe9b230f6617c9fcf20e712621f90d506435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:24:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26830c880e044dcb8575a46b008bc93a : SUCCESS in 9m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/810dc77d0d7b429db24690f163847cf0 : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea42c69cae3f42348de482086742973a : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa76876d10054bb384c09327552215e3 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ccd39f042ac4d8abc1ccfeb8230d9db : SUCCESS in 5m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/881e6479239a4dc98e017f796850a9d0 : SUCCESS in 53m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4533ab8ddd8345baa3bebe3c335cf3e9 : SUCCESS in 1h 41m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c9eba7e50e1f4519b0a6ed712a412cfb : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b2fa23820f3edb0f5ae88d67f2917b934577c9fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:25:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a56daab90dfaed4917474dd43a38356462487656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 19:22:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdb7f6e8ae747f991e77162e10a71b0 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/143cc5374852488ba1fcf0112624032e : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/868c2db69cf54239a9013f8f42eee9e7 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf7a5225580b4abaa2ee501d886d28aa : SUCCESS in 10m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/029aba5155cc498fa41a91e46e95e523 : SUCCESS in 8m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86f8c2642bda43249f36819aa9145ce7 : SUCCESS in 1h 15m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/40efc7d55eff4ba889e8186f42c145d5 : FAILURE in 1h 25m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fcfcd1f39ec244c4a506ae411391199c : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":1},{"id":"854a454d4765a7ab9b8e32b470f6ff7c09f2a06d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-11 03:26:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d73b1bac91a98804de98b57e7d3af824ec6397ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 05:05:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad13fb08843e42e8a3bfc71593711507 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a6848fc7a1e4f319799d16cba5ccfc2 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70e8a210a4474982beadf1194475cbd3 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/244c1b802da64ea591e94fee532f8b79 : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f50d1022f1f4b2eb13d3eec7537a5dd : SUCCESS in 3m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/09a95f72748a4e71b7b40953c0265dc7 : SUCCESS in 1h 17m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6e3e0d1803024431a94b09ea045eaaec : SUCCESS in 1h 36m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7af42cf1a54d4a7ea1ebb2fa903d3714 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":1},{"id":"329072da6f3ec359ce653b97a51abc2271cbdefb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 05:05:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f125b6c2b19117cc04e8f87d35c6cd0a7568d54c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 06:14:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2944de3b8ab65d5ffba31b6a42162a753d47f983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-11 06:14:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa83c24b3194472495106732c21d55cc : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/024b493bed23480aac1fa353afa88823 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77d69c54e19e4dd9847a24860862b5b5 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b797c7c6d0a478d9707c5c702241914 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9da0c21fc9864afe950ffa50bc717f95 : SUCCESS in 3m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d3eab232e9549218fb35042946ea4ac : SUCCESS in 47m 54s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b4dfa762ed39423fb1bea7ed6d16f9df : SUCCESS in 1h 07m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/901f2735a6f74e0384a95a780cb41bf1 : SUCCESS in 37m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41bbda86a0e98cec6e3cb0ce14e99b9501ccc78e","revisions":{"41bbda86a0e98cec6e3cb0ce14e99b9501ccc78e":{"kind":"REWORK","_number":1,"created":"2020-07-10 09:06:54.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/24/740424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/740424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/740424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/740424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/740424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/740424/1"}}},"commit":{"parents":[{"commit":"77ac7b164f7d294ba80832324f4463101cf3fad9","subject":"update constraint for oslo.utils to new release 3.41.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/77ac7b164f7d294ba80832324f4463101cf3fad9"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2020-07-09 13:43:53.000000000","tz":0},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-07-10 09:06:54.000000000","tz":0},"subject":"Fix failure with setuptools \u003e\u003d 49.0.0","message":"Fix failure with setuptools \u003e\u003d 49.0.0\n\nRecent setuptools include updated vendored pkg_resource that changed\nthe exception that we need to expect from them. Leave in the old\nversion, too, for backwards compatibility.\n\nChange-Id: I5bb01fb21982fb8757e925659ad206a15a79f799\n(cherry picked from commit 83b781ae4f4c27c2857439bb02a8c2009a4ee87d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41bbda86a0e98cec6e3cb0ce14e99b9501ccc78e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41bbda86a0e98cec6e3cb0ce14e99b9501ccc78e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
