)]}'
{"id":"openstack%2Frequirements~740525","triplet_id":"openstack%2Frequirements~master~Id1b8cc9aa7709905aa1966f8e9db79ec0a6f104d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id1b8cc9aa7709905aa1966f8e9db79ec0a6f104d","subject":"update constraint for stevedore to new release 3.0.0","status":"MERGED","created":"2020-07-10 20:09:03.000000000","updated":"2020-07-10 23:57:16.000000000","submitted":"2020-07-10 23:56:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740525-1594425361403-64565078","meta_rev_id":"08ccd6b604fbede202fdfca31678e9f60c2a64de","_number":740525,"virtual_id_number":740525,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 23:56:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-10 20:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-10 22:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-10 22:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-10 20:09:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-10 22:32:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-10 23:56:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d3c7053a2460ac9996a5f8091a0c47e05305c42","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-10 20:09:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"815b63b39da995daf3b8801d12a3341a942c8c4a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-10 20:09:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"312dce7d24e6192b4f52d7eb42db2bfec8446fc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 20:51:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cca2f5dc4f4b4c30bc7beda6af59abf3 : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":1},{"id":"92bc37975bd0219c0e27a303e26f851de6f5a530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 21:35:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ff08b635394599885a3544d03e4e94 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87fe1c49b26344328c967610bdb78e8a : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/064cc9944f5841eab9071622f36710bb : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22c8bb34f3854fd58ee12f72f4fa4c4c : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c4ee7a4d02a4612a59a578d50dbd8e9 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e125afb93cae4803890710f288027250 : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7b615d848d324ba38b6609edf8ab4397 : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ace3d27f123d487991e65f2d04aee106 : SUCCESS in 11m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b8bcaf2d37884a3f9b6f7403b4b46d43 : SUCCESS in 8m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f671ee210fae40f79e3721ce071a91f8 : SUCCESS in 5m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/42dafba697d5486dab44d2c80768c528 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d653361ccdf84eea8b5f1858565e3d84 : SUCCESS in 6m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/69d2ea6252e448ad8aa582c0e5e95869 : SUCCESS in 7m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0e120ce5415843d9b935e303ee36641e : SUCCESS in 14m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0c9c4bf113ba4d4ca5f4f7e9d2280557 : SUCCESS in 23m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/244e8d06e93a4336b18d472a26b7fbaa : SUCCESS in 13m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/def1525d9184490cb1b482492b009e8c : SUCCESS in 17m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c6e5267edc484b28ab229057100cb8fc : SUCCESS in 4m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e53c9c9196c4a75940acf5925bab9dc : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/68652fc63c194c129ddbd69bd1c5659f : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf74e8dfa1324b9ca6d4d7cfd32ba489 : SUCCESS in 57m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f85343d376af49b9a89582aae4e9ebb1 : SUCCESS in 1h 20m 50s","accounts_in_message":[],"_revision_number":1},{"id":"61da77cddd2ed9c77a9a5e63cb8a8ecfffdc7ad4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-10 22:32:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0d2128453f5a9ecff773e77c9ef29a38a9174f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 22:33:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3cc4f5921ec82d4abfca2737e34d13d939010cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 23:56:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"63ac96c190b7aced21eb4e660beb010c4963ff27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 23:56:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869adc31c439434d9b27dd251e5738db : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec30cb9a6f6148c4a8d4a0a61529ad48 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f1e728073ce40d59a0e900f62cb6f09 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d030de59283a4ab2b0285b11cd34183b : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1f0ec4177ef4b238edbf16512c42a44 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c8e6a6a19544729954e55f3d75b59b8 : SUCCESS in 6m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6a89a6b5b3348d584935035733ea1db : SUCCESS in 10m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6851a15eb9294636a5c265e2ed68abed : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d02e6d792484ea88bb9a0dc37f3ba58 : SUCCESS in 14m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a5e5ccdf2a03456480ae0760e609d0c1 : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/282a12bd44004b78ba25099f735f56dc : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42ef7d41a6724d80809a022df9984184 : SUCCESS in 9m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0550a30fae0842ebaa7d558f36a3c740 : SUCCESS in 8m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53f464f6482a42d4940a40a1a74d58de : SUCCESS in 13m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/72939413fed74c0e8cdeacdb81718550 : SUCCESS in 25m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6564ac2da10843c9a557272d5afa9cf4 : SUCCESS in 12m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7b26d6b73f5741988f2b84fff5c233eb : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/47355dd954d3448488282bf47e82f78f : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/705452383c54446895ee271a722146d9 : SUCCESS in 9m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0086ece0a54d457fafce7019a9cad9de : SUCCESS in 16m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dfa1197f331d4be1bec6b5f99e0fefb6 : SUCCESS in 16m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e705cee519048599e182fa36113cfdd : SUCCESS in 49m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7547b92a321e4ddeab893683eb5128bd : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"716cb8467ba9a9247b18473fda8cb8c3f1c58a4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 23:57:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc0f2aa79594c80aa54bfb92ddae590 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a8a1a799eeaef2bb862cce270da325cc69dbe62","revisions":{"7a8a1a799eeaef2bb862cce270da325cc69dbe62":{"kind":"REWORK","_number":1,"created":"2020-07-10 20:09:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/25/740525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/740525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/740525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/740525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/740525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/740525/1"}}},"commit":{"parents":[{"commit":"e3dc2f5b268f677497cebe22607697440ce997a3","subject":"Merge \"Add zstd to the global-requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3dc2f5b268f677497cebe22607697440ce997a3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-10 20:09:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-10 20:09:01.000000000","tz":0},"subject":"update constraint for stevedore to new release 3.0.0","message":"update constraint for stevedore to new release 3.0.0\n\nChange-Id: Id1b8cc9aa7709905aa1966f8e9db79ec0a6f104d\nmeta:version: 3.0.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I901c31fae1ce870819682b1eaf34d6768e61d960\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a8a1a799eeaef2bb862cce270da325cc69dbe62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a8a1a799eeaef2bb862cce270da325cc69dbe62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
