)]}'
{"id":"openstack%2Frequirements~741091","triplet_id":"openstack%2Frequirements~master~If0a01e76b62b8ab30b9aa9e89236d2b6f399fe63","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"If0a01e76b62b8ab30b9aa9e89236d2b6f399fe63","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-07-15 06:20:10.000000000","updated":"2020-07-16 18:43:17.000000000","submitted":"2020-07-16 18:38:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741091-1594924723478-4356c651","meta_rev_id":"14363fa5882fa919d4f41f44ca09e74c3566077a","_number":741091,"virtual_id_number":741091,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 18:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-16 15:39:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-16 13:55:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-07-16 13:55:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 01:04:22.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-07-16 13:55:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-16 15:39:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-16 18:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89b7de5192aa53208c67d89744b7d6c163ac28b9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-15 06:20:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"738bfc7f9db8eacff0460da3594d0f3e5a04a45d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-15 06:22:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9512f00d50a71339893c7a9f363a2b69f215ae36","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-15 06:22:22.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"4e8b87817a16fddf7e8b210eed79c128f9bd52a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 06:46:42.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2724b5a5a166478a9aac314d411fb6ec : FAILURE in 19m 59s","accounts_in_message":[],"_revision_number":2},{"id":"8a52ed7aed1688c0ebe441b257126b7e8aec115c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 08:20:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dce379cbfaea443a958caa2361fc2793 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9fe955c27d04e8184ebb4bf54667655 : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f67493b74ccc48bca47274f8085bb2ca : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2569be21d3404e5dbd9f7a5bbe14f96f : SUCCESS in 9m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9a10767348e46d599178b169fb29aee : SUCCESS in 7m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/28f4877abda541f1abe06cfbf1164b09 : SUCCESS in 6m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37576a425d774cc4b989fb9c0d9290f4 : FAILURE in 7m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d3d50c8332b14d9f9f34cf67e46b2b44 : FAILURE in 7m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f6b97bebc6564068ac4de93721465b21 : SUCCESS in 10m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2848cede25234f20a2de47d5f3b4e467 : SUCCESS in 9m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2713b40dd28648a088f29cf97c6956e2 : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/90fcb9ba877e46dbbde98fa703321050 : SUCCESS in 11m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a6f5c53d2c334705b24b49632851ff48 : SUCCESS in 14m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5b3cbbe6be6d4d2fa942543a999bc2b6 : SUCCESS in 15m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d30e27275fe04e868fef5c013b33d37c : SUCCESS in 28m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/060588c40b8042229e3c8678a79eafe2 : FAILURE in 14m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f52a6fbb4c942be82bf17374ff64a86 : SUCCESS in 19m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f5d3a9b73de84441ac6d54ee8d26a6a5 : SUCCESS in 9m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83ae06d4f87547c5947d67e29470ecdb : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/eb13d38d2b5544668012bf1bfe23214b : SUCCESS in 15m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73de93dd3e274b68bd4f246e3460e416 : SUCCESS in 52m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/337230735cce47519b72416ef7972b30 : SUCCESS in 1h 13m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5c2ac696ec0545df6422250b1d5b8981448a4d70","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-15 18:31:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"110ada1e8028fa0224746a299103901c04dcd7d0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-15 19:01:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9ebc636974c3bd887a5aea91e420f19d25290328","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 19:39:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8b82e39b5d1d4486bf6d659c7e43991f : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":3},{"id":"80a39fc9662e42460ad31b74b7df13686c5b9275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 22:17:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a179c2088f4f0b86fd76caa7174b81 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d28200b432341a6ad29571cadb48950 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c92b864e71d43bdbedaeda728b88d99 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f6fe6d5041741fb8b30517316b933f6 : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8de76c1a49274ee0bda07af6c902d84b : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2173c461b6554c59b69dad10aabd782b : SUCCESS in 6m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bf5628a9331f4b1c96f00f22a1526211 : SUCCESS in 13m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f29c8b5c911f426faae33407b33d42f9 : SUCCESS in 11m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/11f8d6a55f46432cb57df6563bd7db5e : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dbb284cfeccd46429186f2c77596b50e : SUCCESS in 9m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5983030f7afe4eb089d53195974839fc : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/939519d19e21440f9b02166b8931635c : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/52c8db6e02f34f3e9360ab238a50e9fa : SUCCESS in 9m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cf509a86dd37457895f48e78365df108 : SUCCESS in 14m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/23b3cdf80eb942998c0daabcc45774f4 : SUCCESS in 26m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/31825a9035a946b0b6f604bab6aab5d7 : FAILURE in 14m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df6aac79d8f04d14a399f5eb628f4239 : SUCCESS in 20m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8946ab70c8241968410044649523b54 : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10d9779ad2a5492996d7d843d558ec79 : SUCCESS in 9m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c80678eacea54588b83cab641042f7f5 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4759f47731dc45da981b9fbfc7867d51 : SUCCESS in 56m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b18769c4420b431ea56144cc483bc6a1 : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":3},{"id":"7f38d6bceae53769c4aa4d330c2d8fc125a462a2","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-07-16 01:04:22.000000000","message":"Patch Set 3:\n\nNova patch proposed to fix this:\n\nhttps://review.opendev.org/741351\n\nNote that I found that mock is calling a class @property itself while calling create_autospec, which seems like a bug, so beware the possibility of other projects not covered by cross jobs being affected.","accounts_in_message":[],"_revision_number":3},{"id":"38b0a0dd02af597941cb66e33770161244ae55e4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-16 13:49:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"66bdb77ef2f2d97ba580d84e12bdc48ef50b4efa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-16 13:55:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0daad74392737b1e0bc27dc83886a4badc1594d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:32:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9cdab0a9702744cb9de3fd971a71fe6a : SUCCESS in 39m 38s","accounts_in_message":[],"_revision_number":4},{"id":"9de58d2779f7798bb4d5bbc7b3d6ba1ef813e987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 15:39:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3af3397a14542389d2458998cd9539d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1665907b2d1e420b8d0e16e7eef3f3d5 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0a7ffad461346a7988ca720bd76e532 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22a3e800620144a09247c96a52d67390 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c109e8182d5451facdf840d37d6bc8c : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f043ca23df114ddb9cfa0c2469263332 : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/821f8a477b4f4144a95734ecb8d549f8 : SUCCESS in 8m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d765c91ed676499c99c09978242dabc7 : SUCCESS in 10m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/114c86b39f2a4aacb34e10f503c8699b : SUCCESS in 10m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d1fc6765eb6d420fb4385e0bfb892e65 : SUCCESS in 7m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3466476b3b3644a6aa1cc78cc4ccb5d5 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c13d6b45cc449d38d7db4f0b8872492 : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b1658adfaf7e4909a82d7d5588035c54 : SUCCESS in 8m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/256e505df3dc4d1c86473c2e6f76b6b8 : SUCCESS in 15m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7bfca904858243c3b68264d084d5ba83 : SUCCESS in 29m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e9a93dc998514f9f9216fb334e4fb280 : SUCCESS in 15m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c62a062339394da88a2daebfbdb11be5 : SUCCESS in 13m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d7f1c27e10ad47bd9a9316f543c07037 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc8b251e6bd84196896ed04a4e71571d : SUCCESS in 7m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e7490043fab142fca5cb00f736426d15 : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9493ad4444d401494ef75aeda313150 : SUCCESS in 54m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff231ecb4b184e36a448ae73d4904824 : SUCCESS in 1h 28m 19s","accounts_in_message":[],"_revision_number":4},{"id":"0a613532f65835e9e7411c80070b9d626ff1442f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 15:39:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"971066a52fd1407bc0112f6cc69d500231adbd37","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-16 15:39:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"39d88eaf4ee8c8003af5e5b72874e327680e27a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:38:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36062ccec9dd4ff8b1144a5e136cff90 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9260c2a7d8284858a87d36b2effa2ee4 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc09eff4befb4b6cbc0bf3274407739b : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8020109c2487477dbedaeb6207f103f2 : SUCCESS in 6m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e70338e9cdc4953b921270c34c2908b : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d4c873dfddf344e885a35792a1561960 : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ff13d2994e2641c5bd5a1e7a175b0a50 : SUCCESS in 8m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b30a6ddb92cb44b08f11fb525326911d : SUCCESS in 10m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6ae11b17bbc941f8bef44943624d518e : SUCCESS in 10m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/367102f6572f457380a8891b6e9cc66b : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5e9c8cfdb1144bc29674987a6411c26e : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6256bd5a2b554346b41de53b02637c6c : SUCCESS in 9m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/52a899c2541e4a31b0500e4b62ff48c1 : SUCCESS in 13m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e85307858c5643098d58c0e7b380093c : SUCCESS in 13m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cfff5db8872745b99b4ed5776385202c : SUCCESS in 27m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df22695dd8ff4c5f93dd15d79f925355 : SUCCESS in 13m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e9dab42757d44c9da1ed79cdc195ed36 : SUCCESS in 19m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/af1e0985e37b4c75812055d07c998a11 : SUCCESS in 7m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f42d0d32ca7d4552a14d0141cae0726a : SUCCESS in 7m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cd0a0c7075e845769530263492681d53 : SUCCESS in 13m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a09cd6efd7d4863a903be7f4426d6c5 : SUCCESS in 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62d453d4f5154c0880280135c4b3aea7 : SUCCESS in 1h 07m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77669bae56a84c5fbd1a41ec2a1e75dd : SUCCESS in 1h 23m 54s","accounts_in_message":[],"_revision_number":4},{"id":"8c83b1b15cc1f72057db21944221d368370862a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:38:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"14363fa5882fa919d4f41f44ca09e74c3566077a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:43:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e8b7091a3e46258e1537d4ef088296 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"26c10f7335dfa6e2cb9274578b82f4d5b03bc40c","revisions":{"7efe699087eb6e0e327bb2324eb8b02ef53aa06e":{"kind":"REWORK","_number":1,"created":"2020-07-15 06:20:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/741091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/741091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/741091/1"}}},"commit":{"parents":[{"commit":"0d27f316842f38ff558b9bd09a4b48ddb7472f74","subject":"update constraint for python-openstackclient to new release 5.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d27f316842f38ff558b9bd09a4b48ddb7472f74"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-15 06:20:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-15 06:20:07.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0a01e76b62b8ab30b9aa9e89236d2b6f399fe63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7efe699087eb6e0e327bb2324eb8b02ef53aa06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7efe699087eb6e0e327bb2324eb8b02ef53aa06e"}]},"branch":"refs/heads/master"},"31d52b852beffafb322f899b29fedf3b190f61dc":{"kind":"REWORK","_number":2,"created":"2020-07-15 06:22:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/91/741091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/741091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/741091/2"}}},"commit":{"parents":[{"commit":"0d27f316842f38ff558b9bd09a4b48ddb7472f74","subject":"update constraint for python-openstackclient to new release 5.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d27f316842f38ff558b9bd09a4b48ddb7472f74"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-15 06:20:07.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-15 06:22:04.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0a01e76b62b8ab30b9aa9e89236d2b6f399fe63\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31d52b852beffafb322f899b29fedf3b190f61dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31d52b852beffafb322f899b29fedf3b190f61dc"}]},"branch":"refs/heads/master"},"55e7c6065fc3d452bea82da9a9603a8ce9ca62ef":{"kind":"REWORK","_number":3,"created":"2020-07-15 18:31:43.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/91/741091/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/741091/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/741091/3"}}},"commit":{"parents":[{"commit":"0d27f316842f38ff558b9bd09a4b48ddb7472f74","subject":"update constraint for python-openstackclient to new release 5.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d27f316842f38ff558b9bd09a4b48ddb7472f74"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-15 06:20:07.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-15 18:31:35.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0a01e76b62b8ab30b9aa9e89236d2b6f399fe63\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55e7c6065fc3d452bea82da9a9603a8ce9ca62ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55e7c6065fc3d452bea82da9a9603a8ce9ca62ef"}]},"branch":"refs/heads/master"},"26c10f7335dfa6e2cb9274578b82f4d5b03bc40c":{"kind":"REWORK","_number":4,"created":"2020-07-16 13:49:33.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/91/741091/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/741091/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/741091/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/741091/4"}}},"commit":{"parents":[{"commit":"0d27f316842f38ff558b9bd09a4b48ddb7472f74","subject":"update constraint for python-openstackclient to new release 5.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d27f316842f38ff558b9bd09a4b48ddb7472f74"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-15 06:20:07.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-16 13:49:21.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0a01e76b62b8ab30b9aa9e89236d2b6f399fe63\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26c10f7335dfa6e2cb9274578b82f4d5b03bc40c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26c10f7335dfa6e2cb9274578b82f4d5b03bc40c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
