)]}'
{"id":"openstack%2Frequirements~741640","triplet_id":"openstack%2Frequirements~master~I94e44017ace431d90d57fc51e82abbe5232cd7cd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I94e44017ace431d90d57fc51e82abbe5232cd7cd","subject":"update constraint for castellan to new release 3.4.0","status":"MERGED","created":"2020-07-17 13:52:25.000000000","updated":"2020-07-18 03:32:12.000000000","submitted":"2020-07-18 03:30:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741640-1595043035537-f352bdee","meta_rev_id":"2f8723397a4217a3dbe4db185e318baae2c54f52","_number":741640,"virtual_id_number":741640,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-18 03:30:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-17 14:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-17 14:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-17 14:14:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 14:07:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-17 14:14:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-18 03:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61de4d03301a2c73e6cdda4e83bfd33036f7f173","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-17 13:52:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"482b8fcd369c74c4c0792655cb042202ef8f0f25","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-17 14:07:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b38fc19de9a3852745195c1eb0879e8cd5636e25","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-17 14:14:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10ef125163e64d6c51184c74fafe35d50238a061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 14:40:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2e3337ee08a748dba0eda7bfef58a59e : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cbab3b288fed06412767db35d2ac672876e306d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 18:37:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46d35ef3404417a9f164923564ba5bc : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad23a057d40424da6365f70d682f4db : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdf954496c1745ac916ddb881d1d723b : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42735c26ffd74bf8988f834878ab3039 : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/753b8e5805de4940be1be10e1319afb7 : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be617ef70fcd4ac9add4f5258aeccff4 : SUCCESS in 5m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/74f0aa03a3ca4da0955990d8abc298d3 : SUCCESS in 9m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/39d92b54ced24c1a8107d6ee9a63811d : SUCCESS in 11m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/93cd78ef95164cee9d3839cab8803d88 : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c64f0d36e89f4ec48c84c8125c687a41 : SUCCESS in 6m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e06899dc7257475f8a928cb9d808970a : SUCCESS in 12m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3bf2bbec9632437d836c62405d1b26b9 : SUCCESS in 10m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e3c1e4392c0f4ad68384ec938ffd1b3b : SUCCESS in 12m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b833e633930c45f3ae0774fd17495018 : SUCCESS in 14m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/055bb55eb68c4d08afcc507b650f3482 : SUCCESS in 21m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/041397885b01483a814a275f090447f7 : SUCCESS in 20m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24bb1e81eb914b48a08fbd44588344f6 : SUCCESS in 17m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e44a3065996d4af6a1e3b3eabb3b3c8b : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86f139e4c3e54d15a51c914572ee5283 : SUCCESS in 9m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bf653c1eb6804171b2bfaa31401aab0e : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de81b715babb4a8b8f4ad33c7df38992 : SUCCESS in 53m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/978621202afa4d00994e078c76bb402e : SUCCESS in 1h 21m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6f8c6949abbfea8df257f3e53362e62d49bd70e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 18:38:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b442c19c375a9729a246fc8048a88aa94a29bfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 22:58:58.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae67a0d1192b9cae3759f71904ddc777bbccdbf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:30:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"797f6c93e47eb86f6ec219c0f0204f1bb788a22c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:30:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6d31179b7e47db8a1dd8be592d1a75 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c452217e426b4a3db1fa07cc3c29696f : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c3defb6d1804c648c6af29893ae4dc2 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77c68abc52df42da9156e258056756bb : SUCCESS in 5m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed1cd58ce991405e824a7d70d3d84c07 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9822410f4d08427181f7feef2e4cd7fa : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/416f098d03044b83be8c2f17b7afc84a : SUCCESS in 8m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/88640736286b4c1db370d2681b8d1455 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e0df7e3f46a84a3ca4ff7dedc772030c : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/309cb8329d3e4b689179d1898c2d4d1c : SUCCESS in 6m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bf2e51fe70be4fae97699440419a59cb : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/457f63c361314300ac39e272bb403d48 : SUCCESS in 9m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c2abe1baa03e4fa3917ec217e04c2afd : SUCCESS in 10m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6cfca23f34644b51b4f5be6d9f2921bc : SUCCESS in 15m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4146e0b1b4154b52ad03e16e4604db9d : SUCCESS in 24m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e636782f4026400c80c1786fd0ec17d4 : SUCCESS in 16m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37b3e9deedce42aea7ed13aeb7c28d6b : SUCCESS in 19m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6659b49ccb9643918852cd944f2abf4e : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f85a1d4bc05d4b94a27b529cc09574a8 : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dbc7cafa258142e5b43f02757f41b4e1 : SUCCESS in 11m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/713e3b760ce747f3a65a8662bccaf44c : SUCCESS in 10m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f4031f9ae9948e088ef556ccf93f7b6 : SUCCESS in 55m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb6ea19f858245b9b964f31c4068ef37 : SUCCESS in 1h 27m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a27afb36905ac66303738f2bf47391f30586122a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:32:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a2d60b605cf445d8ed2b2f82941f29e : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"378903a03fac582f13c2d2a9e89aa26beda98f13","revisions":{"378903a03fac582f13c2d2a9e89aa26beda98f13":{"kind":"REWORK","_number":1,"created":"2020-07-17 13:52:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/741640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/741640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/741640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/741640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/741640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/741640/1"}}},"commit":{"parents":[{"commit":"ac05a3b77354c40c91c28239f9584889c2b1501a","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac05a3b77354c40c91c28239f9584889c2b1501a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-17 13:52:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-17 13:52:23.000000000","tz":0},"subject":"update constraint for castellan to new release 3.4.0","message":"update constraint for castellan to new release 3.4.0\n\nChange-Id: I94e44017ace431d90d57fc51e82abbe5232cd7cd\nmeta:version: 3.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I12855da61e2466a42bb81c18bcc9851e52eab536\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378903a03fac582f13c2d2a9e89aa26beda98f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378903a03fac582f13c2d2a9e89aa26beda98f13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
