)]}'
{"id":"openstack%2Frequirements~741645","triplet_id":"openstack%2Frequirements~master~I2ee8470d6f8bacee3a3adbb2449c10ea8e80c475","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2ee8470d6f8bacee3a3adbb2449c10ea8e80c475","subject":"update constraint for stevedore to new release 3.2.0","status":"MERGED","created":"2020-07-17 13:57:38.000000000","updated":"2020-07-18 03:30:42.000000000","submitted":"2020-07-18 03:30:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741645-1595043042418-f352bdee","meta_rev_id":"6a5c84ff0af80fd1b887c3418dcefb103944148f","_number":741645,"virtual_id_number":741645,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-18 03:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-17 14:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-17 14:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-17 14:16:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-17 14:07:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-17 14:16:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-18 03:30:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e943fa58fa5bf628f8396ea700b4d6d12ed0fe2a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-17 13:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e991cb306428dd25595c236f5c6b3b3626e67c51","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-17 14:07:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22bd2d789aaa1a0c675d813691da1d0994fb5f9b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-17 14:16:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86e6c0dbece25fda10722b901cd439028c71009e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 14:46:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/791df5fe1a0f49348a68a8d04c4a0dad : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b09ea2d4a5cfef479aae9064030a2a9c6f359b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 19:55:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b25009a18445618b3274b0915bca1a : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3110fb3e8994e80877cefa4e25cd95f : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e16d4d6f3ba4522a76ed6cb06141dc5 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1bfbaf46bef424e84fd1d703e3b232f : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2be631faca124b2fb91027fb14737a2d : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8042f09c24e4fedb6d49049de10cc78 : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4535822725b9448291d2f3be8921193f : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/707f95051a694919b8820a518c42e7b9 : SUCCESS in 11m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8e2548c92d894fc5bd19f16f8770a06f : SUCCESS in 12m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/213e551faf62484c859deab94ee7b2e5 : SUCCESS in 5m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5d1e7c4a59344f13ab9765c4b25f830c : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4f6bfbd82eb4fb7857e732ccb50e789 : SUCCESS in 10m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/29aca9c44c374bda85071c58e06e70bb : SUCCESS in 10m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b0f39dc372cb43d2b6f9d702b6987963 : SUCCESS in 17m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8a8199bdec9442fdbcfcd5968434f21e : SUCCESS in 29m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f015b52596a7440dbeda1b75b5108011 : SUCCESS in 11m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8402f54645b44af8ab434892261bc20b : SUCCESS in 13m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6884e1ebacd14203b2e75e35611eff15 : SUCCESS in 7m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c43358efcce147f79f9b9d2af38bf1a2 : SUCCESS in 6m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8cd82f0f28ad44cab43ddf95f5a7fb23 : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5a7cc122cd441b09f5df75b291ad11a : SUCCESS in 52m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6e3eff8280a42e0a1f7127865ce4275 : SUCCESS in 1h 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3759bbcd323d4796ec10837f44d55ef43ab90f25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 19:56:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6fd50eca675869d05a53190ea68180bde0334375","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 22:59:14.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e4bf063139b8826fa7059d59681c54152efa679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:30:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446e4d8bfb1d421ba555a6fe244db8a3 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba312b789357417bb7be562644ee426a : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9def0fa02fe64ca9a378f01697949f8b : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25cfb4b5b0c54648a338a7d37516f9e1 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c3239af4214349cfbb59c29a39075269 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f7db3ca6c8b4f2db14216ecfb535560 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/04b77710e8de4b2793cdf4e76fc33f87 : SUCCESS in 7m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03f6c31d82df4acfae6a1e5f1ac24f19 : SUCCESS in 11m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/57a5abf44feb42b1bb5ab8e6b77534c4 : SUCCESS in 13m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6e664bbfb4bb4145b7c4a7deb29be83b : SUCCESS in 7m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/59791a57fe214ec4b27cf142ca273bc3 : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e6656968b8f4417a08b417f70cdd11e : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/115038b878c24dd3aceaf98be72bc299 : SUCCESS in 11m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/02f7b54f89324382b9091813d930a2a2 : SUCCESS in 44m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/15a2184479f7482c85ca9d8918360d1b : SUCCESS in 17m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/98ae714e55574ca2af803eb18b61fc0a : SUCCESS in 13m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c6ad271d3b2542edb221ec25c0e9c6d2 : SUCCESS in 17m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9eac4ad9f09541bcbe1da537f9faec9f : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9cafb3966fc45a395a0ecd8fcd89608 : SUCCESS in 7m 58s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/04da18be6e934e43a94723e413a6240c : SUCCESS in 12m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/63aab2a508ad4d12a551fce71998d19c : SUCCESS in 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb429325eb7a477aabd803f6f494cb4b : SUCCESS in 49m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e3a04a9d57f4a08af66e9eace91e824 : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":1},{"id":"cce698d40503bf1852dc8694e7ef216014bda871","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 03:30:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90680158354e61b702ceef0e7065068c5c570cb0","revisions":{"90680158354e61b702ceef0e7065068c5c570cb0":{"kind":"REWORK","_number":1,"created":"2020-07-17 13:57:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/741645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/741645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/741645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/741645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/741645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/741645/1"}}},"commit":{"parents":[{"commit":"ac05a3b77354c40c91c28239f9584889c2b1501a","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac05a3b77354c40c91c28239f9584889c2b1501a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-17 13:57:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-17 13:57:36.000000000","tz":0},"subject":"update constraint for stevedore to new release 3.2.0","message":"update constraint for stevedore to new release 3.2.0\n\nChange-Id: I2ee8470d6f8bacee3a3adbb2449c10ea8e80c475\nmeta:version: 3.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I12855da61e2466a42bb81c18bcc9851e52eab536\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90680158354e61b702ceef0e7065068c5c570cb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90680158354e61b702ceef0e7065068c5c570cb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
