)]}'
{"id":"openstack%2Frequirements~741837","triplet_id":"openstack%2Frequirements~master~I21039783ffcf0bb0257ecbaeeffecea3fca579cc","project":"openstack/requirements","branch":"master","topic":"pymysql-0.10.0","hashtags":[],"change_id":"I21039783ffcf0bb0257ecbaeeffecea3fca579cc","subject":"Update PyMySQL upper constraint to 0.10.0","status":"MERGED","created":"2020-07-19 06:22:46.000000000","updated":"2020-08-11 18:15:09.000000000","submitted":"2020-08-11 18:12:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741837-1597169554622-ab5395dc","meta_rev_id":"83610c8c7818ccb241ba5b6ce9114a48bbf61551","_number":741837,"virtual_id_number":741837,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-08-06 16:52:34.000000000","_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-11 18:12:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-06 16:52:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-07 16:19:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-11 16:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":1,"date":"2020-08-10 13:57:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-11 16:40:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 11:14:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-06 16:52:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-06 16:52:34.000000000","updated_by":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"reviewer":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"state":"REVIEWER"},{"updated":"2020-08-10 13:57:25.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-08-11 16:40:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-11 16:42:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-11 18:12:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5cfa4c16ac84d5f8dd15d8bbacccb1578f410cd3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-19 06:22:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d27ade2b4e9d2ccf5c15e40fe9584dfba3a47809","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-19 06:32:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9ea528fb15feb950ee7dae60a433145a553c29f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-19 06:49:10.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"65db77796bd4b279666a274c5557e429145dc9ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 07:30:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/def83566422d4cdd99d10c85c50a8c38 : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":2},{"id":"5ef199174b6807f889b1f49538939979b1c34e96","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-19 07:39:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e38e807ef7aa3690b7e61bc7bff5166bb29bd41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 07:58:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083494566b014782b1639aab53a8d1fc : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24857f508e14700ad9f02bb8cd48dfa : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05becae362e14ad08f5b930e214a3146 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84f685403798401aa7c1c3e2a039cce2 : SUCCESS in 8m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4075947b11184ecb8a27e9e31923374e : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1f5eeb101fba4c43a63087e2f412d8aa : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d4235e347e92474097fc82eee1f5bdd8 : SUCCESS in 9m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cce5cd1a914e4c6ab799a644b38acc81 : SUCCESS in 10m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d7f72df026524ba6979fe0e696e31f7f : SUCCESS in 10m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/673a458fb8e24617b8464cab08aedea5 : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d0289aaba5e84197a25c406587f6b01e : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8dd412cfc6fb4699b68e374da211e1ac : SUCCESS in 8m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e29fcf241d8945b5a2853aeea3438a54 : SUCCESS in 9m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/59e741e8148c4c138c800b29f38d87df : FAILURE in 13m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3e88b3d6470641839d73258417307966 : SUCCESS in 26m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f5730d09fe114cab9ae3dfa95f5dfd59 : SUCCESS in 14m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e983fca6523143b98406461edb2b4293 : SUCCESS in 20m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ca74797da4524eb7b4731cfaf9623ecc : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89094f9649484f7aa7da912985837333 : SUCCESS in 8m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4add0acc18a9459ab8010c5c2b03459a : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de0d55c0fdf04833bd91cddfc1952d32 : SUCCESS in 53m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da9caadec07b4a1cb9569acd33b437d1 : SUCCESS in 1h 15m 36s","accounts_in_message":[],"_revision_number":2},{"id":"52fd8e546188de1689299acf2ba55449a15582a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-19 17:28:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2f52340574a853eaf36e50ec92de2c8401d39584","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 18:06:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9a7a564aa3494ec5897d7a4878c3e746 : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":2},{"id":"18fc716b9e0245116845dbf7f02601c347a3a696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 18:22:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b9ca750e3e4df1bef6714b219f0061 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff9f8ed717364f038c9512ac43ec629a : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7a52fbddd4548239837a41659d41b73 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1f5104da552423083364c280eaf8a24 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0a483bbbf1045fe8f5536c436d5de66 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf3590833ee74bfa889ee2b4e73b76c7 : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f5dc69cdc1b04552a5d8d0dde086130b : SUCCESS in 8m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e6bc17c699143e4a65d1b8220d37a07 : SUCCESS in 10m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/25e01738d8c34b3aa11146f435ad31c6 : TIMED_OUT in 43m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aea5fb3e07d64bdfaeb48b83dff99e45 : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/769b22ebb95f43c787c2111d3e0a1d5a : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a6d208b0d47f4a788f07181f4708b363 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5f965ad69f1d4e2a8913ef27134df4d5 : SUCCESS in 7m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/68e3f52c8bb34c10bcf68741ad59bff8 : FAILURE in 19m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d704b6bb653349ea9762bd544a302dc4 : SUCCESS in 16m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/36694d8e842045dcb43ecbf52479925b : SUCCESS in 12m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f8f2d43d08724fc1b7d94caf4bca3d7c : SUCCESS in 15m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df63257cc71449158edfc56e52869f9d : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c4c97d2744747c6b3f47b7bac20dfbb : SUCCESS in 5m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/97741322d70440ea9451c42482820cdf : SUCCESS in 8m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3821056958ae457a83678fe20cd44192 : SUCCESS in 53m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c2c4245123246599b066ccf57021dd6 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a42c89a6a7c03613119727a2c941f0ae8837405d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-20 11:14:56.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7747af88c197ad3d9ec8c5835647d523ff335b3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:01:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f355e838796d4009bc3414d1229d76bf : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b3480de2b22c14b8319f7f7968c9692161ce7616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:55:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1affe783c334d5ebd5524825543486b : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a59c99b7cfa44819e5a70e0541c00cb : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b6589665a7e487e896a500a0bdbe777 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8106b550ee84f91b84229f94c23d6b3 : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/488ef792259246b8a9bcedf3b474c8af : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34a3b56bb61e435ab83dcdabf2938c92 : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9e9d7a2390a42e9ab946b23d04a754a : SUCCESS in 5m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56039b123d2747e4aa0532591b71c70e : SUCCESS in 11m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/021d8a6738d348418273f49c6b22336e : SUCCESS in 6m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e3c0079c01444d83a70e5edf7c14487d : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2b8a208ba7bd4bfc9f621300a6ad643c : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24398a4941674e5ab91a8e61924e500e : SUCCESS in 7m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/479fa9e4b860416fa802cee878ee821a : SUCCESS in 6m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/10339aecc93c4cd6a44ac42ab8c237f8 : FAILURE in 11m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cab898b01d6541128a077b30f45453c6 : SUCCESS in 23m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/664834ad3e4d47deb3268f19114d4384 : SUCCESS in 13m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3f0a24a701d4055aa12d943601c2197 : SUCCESS in 25m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/48abae1faef5440bafab7c7c4c8df469 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/abb9f4995f6a4ded9f66685c839a2e9d : SUCCESS in 5m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/371e3d2bd17549238432176d32cf370b : SUCCESS in 13m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a43b4f142eb4ab68275fe96856dda68 : SUCCESS in 50m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7589e2459bce4955a2e140bb409f9fc9 : SUCCESS in 1h 39m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8cd29faeb9a95d6006b5ce7cf41bdb62f26ce2c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-22 11:52:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7ead68b8a9195ccd351d3e96000cba0b038f63ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 12:35:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b77e68776f964600a057bf56711c77e4 : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":2},{"id":"008919b82e93ece5a03cbdf070df62f7ddb8919b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:26:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da027fab19f8488d99cce1b88c5faca9 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50193bd4029941b89d74f9df2f997e6c : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a6a54065f214510a87272edc67dc3b8 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/877e6731d4ed474182766d9d785a3c9d : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0575039af384483d9f196e6702b9e87c : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dab820f1d2894e0498bac9651bd584e7 : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac1b82ed74a24ca68e9ed190669503ad : SUCCESS in 9m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8f64a317034417aa5c7294e3e45c6c4 : SUCCESS in 10m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/54aaa0913d524de787afe2f3387b30db : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/135e1708273e4182bda73f6fb10e9fd0 : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8e85116dd6544a299ffdcc6e7664caf7 : SUCCESS in 12m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4ebafeb73b934e1fab6e72441d274f15 : SUCCESS in 7m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/16bb0f890bfd45ee9fbfaf201dd3960d : SUCCESS in 10m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/10fd41a125934df79675c83a341b90ef : FAILURE in 32m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d234d7a9cca446649feb77bb5bb77308 : SUCCESS in 28m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cf83952d09ef4da99c9965e97d63bb50 : SUCCESS in 12m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b52c24551222453b91474434136ebad4 : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b726a1e3a5c7404a99c83916dc83d8cd : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f4ef720e1f345b79c6e65c5d54dd4e3 : SUCCESS in 7m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9619dbde4c484731bd7cd7f5b039b739 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b75c73fb65a4aeabc6fa6b64b0df62e : SUCCESS in 55m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c913858b7f34afc8f0a91979e209f9b : SUCCESS in 1h 15m 57s","accounts_in_message":[],"_revision_number":2},{"id":"36972112945626858c2df1db69752006627729ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-22 17:47:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9e8cd3a052208ca16d1d3aea074f8322a13b7f60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 18:27:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a632d7ebdf784457b68673e7ba99a013 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":2},{"id":"cfb487ccf33589cdf69715e38b61c091e600257a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 21:04:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b7318cea1f64072b901839a85fc0456 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c408541a8f4b019df5ef4055d70dae : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/443ee8de37534652b4532d196e2f973b : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6917593aae004c83aeb7e8e6b554bc99 : SUCCESS in 7m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0f5fe0808b24886a4fcaff1ab4a47c1 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/37c97937cf2c4f9e9be95e614a6d9e4b : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f346424612504453aea18e68bbbeadcf : SUCCESS in 10m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3c30fe855d0944dc89a5292269c0073d : SUCCESS in 13m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0772261535564763989b922b5ab11b22 : SUCCESS in 10m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/00bb4e6ad39f4bb69ef504915b702ccc : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c04c87751fef457ca408664b1f1340e5 : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73e5f1cbb1ed4332bad6c28a0dbc1650 : SUCCESS in 6m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/24846ba4ae9247cab4e1ec72d3332989 : SUCCESS in 8m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e7fd6f4e39274aad92d03dc53ddbea51 : FAILURE in 24m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b20d4a7e42fd4134b9890d26b90ee30a : SUCCESS in 24m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/071be81dc8ea4047a715591e3050d615 : SUCCESS in 14m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a319cc30346e4e07bb05f81aea31a3d8 : SUCCESS in 33m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/19de12903da74192b3d8db6cc68fa0a2 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/402b72cc686c43a0ac6134777113ae56 : SUCCESS in 8m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a771a94c37864902a3ede1f630daf61f : SUCCESS in 14m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d532a89962eb4755887147ce2a70e72d : SUCCESS in 50m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01ebd7eeb576479dac764c21328d1d2d : SUCCESS in 1h 22m 09s","accounts_in_message":[],"_revision_number":2},{"id":"6703a2fa45a26dc6c3f411c908fda43cf14a0acf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 19:02:31.000000000","message":"Patch Set 2:\n\nLooks like this new version causes problems with something in keystone:\n\n\"Identity provider migration in progress\"","accounts_in_message":[],"_revision_number":2},{"id":"03893a25d35f0e88fe6c4719b9637e43597147da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-31 21:44:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c023df9d5e0218dbd0037c73ff36ca3149773061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 22:08:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/533278cec7e940a8a4f8548198d1ac20 : FAILURE in 11m 21s","accounts_in_message":[],"_revision_number":2},{"id":"eac356ba6a710e7b57602286a89496ff99eb841e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 00:30:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/961682ccb65e4f2594281830cc8b4788 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aac1d8ee4284cf789abb1fe6a3f87df : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0971c3a320f34bbbb5cb051d25820959 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/245d90cd5b654402b34381163cf6aa53 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fcd35a5f9e54889b51a3fa9d4deb452 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8ae12c132c8402fb3c58e8208155177 : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ec4400899f5f4cf4a63445b14b92c654 : SUCCESS in 6m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f076f9d31eec4ff1b702f5c646dd4135 : SUCCESS in 12m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2989c2db89c54c02b60ff0fd16ce38af : SUCCESS in 11m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ce20aab694564c27af7502cd18724c7e : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d9781902ed354be5857881cda26b77b7 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0286ae5b9c2b4a279b567f2d91ad559b : SUCCESS in 7m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/32257b20d41e4f31986dab0bfc578973 : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/65d5b97d57154e6da7efbfb87b077810 : FAILURE in 32m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ea493c3dbcb44e8e9238a6238be06d54 : SUCCESS in 16m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0ccc5e5a486c4d9484b23513df207cbe : SUCCESS in 15m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/77c7e0c1f9114d3ea0f636a0b75552aa : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ffd2b8174b854e6f9b6c7744b6fc90d7 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/766e5ffe81944435a004fccc10c25546 : SUCCESS in 6m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1e8c01cd8576499fbd9a0956d3609845 : SUCCESS in 19m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a40f55acb99a42059ce3c737d421d27f : SUCCESS in 53m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac5702089fb34142b63818577d15a325 : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ec1dcad67de866c6047d1a84081b6f9fff21d1cc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-01 15:37:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"32eb0eeac950c1c82b0828cd2b710a50f9abcc54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 16:05:21.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3fb0084cd027452faa15da8fff2b5248 : FAILURE in 15m 53s","accounts_in_message":[],"_revision_number":2},{"id":"6f8b02251e4b3129f00a507509e0906d56e8a93e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 17:04:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304d9019c9b14473b8929654501b13bf : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1000bc85059d4505af2a9a5c48210b1f : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/636facc31e244f3292f27ab01b840185 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2141c47b4b1841efba8d37a2f7ec5e81 : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/509b50eb83cc4affa14667214658de83 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c230225129e4dbb9258e7f7ada48788 : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/609dde6ab1f741ad9b59a8506d5ccf92 : SUCCESS in 6m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/280bd3d709b345d195eef7b2dee52caa : SUCCESS in 9m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b17103f7166c4ce7be36099f78219cbd : SUCCESS in 11m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d7591522841c4a4599782439d2708a06 : SUCCESS in 4m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2b459a7ac7ef425fa776068ff19c7d7a : SUCCESS in 6m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a346390bad0749d99a41f04e04f3eb82 : SUCCESS in 7m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/473b1ed0c19849359dde689cf19bf367 : SUCCESS in 6m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d1a2e7354360418b9835fad3e5dfb9cf : FAILURE in 10m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a38234950c5f49e0a57a0bca293023b2 : SUCCESS in 24m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/514e4c7f72b0452f961eb64b5b388ba2 : SUCCESS in 13m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5369b16d126e4e51913c511c9b86f934 : SUCCESS in 15m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d871f023f4d541719c913d621e4d7c68 : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a3e84557e9644b5b54906d12da39650 : SUCCESS in 7m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/04dc80dd674644a8a9e4fd1992657bbf : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad8501688d7e4d17a8666f8428db9dc3 : SUCCESS in 36m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/454251a072c240759cd78fd6226ab67d : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":2},{"id":"830486c0d1f0d6439a346bb84d600d2202e05eb7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 22:32:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"631a291d4ba287764d0c8d9a56e0cee8691ae277","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-05 22:43:02.000000000","message":"Patch Set 3:\n\nheh, fair","accounts_in_message":[],"_revision_number":3},{"id":"14d3a174a305afcf8c753fd19484add6829df436","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 23:00:55.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e9c83ccb6e7e476fb9759b658233a3c3 : FAILURE in 15m 26s","accounts_in_message":[],"_revision_number":3},{"id":"61d680061661b6a2591d2020f5825e9dd1da18ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 23:46:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ce3e5a260a4b398810ffa30a356812 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99fa96a21a314c1684e34054632efab7 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92d7dcdefb4c4df69e7c0ce65f9986cd : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d23d07cc8d74269a15c90b65d33807f : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93de32669d494d1b87500f451940d2fc : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2cd1fb22ca5b449dbd62623c99e8abf0 : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8ac2e4448b6e4d11aaeebeb9b3d3cf11 : SUCCESS in 7m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ba1ad16b0eca4b92bdd0e197a6c8744e : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d0ed6965d6c94460b736c948ff3971a7 : SUCCESS in 8m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b717c7544160480ca0c1c27d3e7680e1 : POST_FAILURE in 5m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/373dd95dd5024c739e026d89425d51b7 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a283611e2a394cce99c067149ad73f63 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c0d3c74675f24480bfaf425a85955016 : SUCCESS in 8m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/94584dc78e1640fd979b1864f4bf133b : FAILURE in 17m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1007f59f53bf4db2893d4fdc817615a1 : SUCCESS in 23m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0247da887aae4dd8b109c71c38f73b97 : SUCCESS in 13m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa81b573d7c5456895327d2d3b114b1a : SUCCESS in 15m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a7d0908ef0ff42d2822a948a181decb1 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/116dc46ec88e4262945ec541036c4838 : SUCCESS in 7m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7d67a77df08146b7b408799c6d4a5847 : SUCCESS in 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ceb654a0c7749de889307b95fe1f08c : SUCCESS in 59m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5e54106ffd041a7ae00958924f0a3da : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":3},{"id":"6439aed09301e85e73758d80aee108ed6f1d039d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-06 16:52:17.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ede236833e5825f2f73bd623e22ecba8ea2ce1eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:52:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"2f6e6db62277827e70dc09ba981534fd2bde8066","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:52:29.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"3c5b6be63fa8cd3a71e13affef3b9284d87fbf50","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-06 16:52:29.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"792ada7b1ef0b6b67549b8b10e602641fa4adbc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:52:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"383c6874bb465244d26f6ed081d6920c0a6e8fba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-06 16:52:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"c6e498efe904b65d069a88c00fb3ba06b5d8ec26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 16:52:31.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"ca4850c20e794abe0fe5fb56a6dcae4fc16cc6d8","author":{"_account_id":28801,"name":"Cisco Cinder CI","email":"cisco-cinder-ci@cisco.com","username":"cisco-cinder-ci"},"date":"2020-08-06 16:52:34.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n\nWarning:\n  Dependency cycle detected","accounts_in_message":[],"_revision_number":4},{"id":"ca9f9b47fa2dd05d76711742066ceb779324c1bf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-06 16:56:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ccee5f0d01b21d97bea437628b55e3ac4eee54c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 17:14:21.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fb93d6766a06438c877924073a9ae161 : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":4},{"id":"8bdc4429bfcd8f9bd51b9204d250cf8ed4e7acc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:17:23.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"aaedb77fd95d33e0f090329511c93c0223fad30f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-08-06 18:18:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2a503c6524bd82a02ec875c1cc103516841bbf47","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-06 18:22:41.000000000","message":"Topic changed from mysql-0.10.0 to pymysql-0.10.0","accounts_in_message":[],"_revision_number":4},{"id":"e75535bc1e875d19978d5b7ab47e91a07938307f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:32:19.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/333b740de58c448d956c06cb72d47f71 : FAILURE in 10m 45s","accounts_in_message":[],"_revision_number":4},{"id":"578478ba10b084dc4ef790e58afe7371d5586539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 19:41:15.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711a75131b124f738ac0d3e4b853f84e : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08b55000bc1f46ebbe397aba21904607 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bec71e6d72e94bfe9eaa041fdffff0d2 : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d649e959dfa94f7cb89a9dfe25156427 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3867b1169be64f2bbbe563baad8e2aba : SUCCESS in 7m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e32566961f124a3899fa5042b033150b : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/44b5a0de50b24ca3a6865cb57af36bf3 : SUCCESS in 13m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/40575fba7cdd47d4af9736c8dd53f8ea : SUCCESS in 14m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b7fcf4c2e9543169d5fbd34414e928c : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/de79b26282e84ce99dda9ee8a5966222 : SUCCESS in 6m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f14f5cf46cc4e9cbb0476a8b751dd03 : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71865ded8b794a93ba504c6439aa56ad : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e10b26f92f0b44c283bc3adfcaedaa9a : SUCCESS in 10m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0d3a1e5d5cba4e3f9ed4633fb26c4be6 : FAILURE in 13m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5e2dad053a524a67bb56bd924e29c373 : SUCCESS in 27m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9dd829acd86f4650a4ca5ad69324235b : SUCCESS in 13m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f79238c215ec400c8155468e18278a28 : SUCCESS in 19m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a8c3f8c3fb874b509eca5dc4d6b011c3 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0c7c13f98de4c28915bb065e542ffde : SUCCESS in 5m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/314b38f53bde4d1f826db3ec1f1eca15 : SUCCESS in 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4e03bf8d7d242b9a9860b784c96dc27 : SUCCESS in 49m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7caee1b4cb624da58fb59bcd4fdd0353 : SUCCESS in 1h 21m 04s","accounts_in_message":[],"_revision_number":4},{"id":"0260a53a10be26b3d959e027ddf94b66e080dbc4","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-08-07 09:25:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"de25d7ebaabc8ea77c2dec60b92730fb2b9aad4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:35:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3f8222287359456ab834b2da2b94092f : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":4},{"id":"ad10d6664585d21340a3bdfce803323ebb8fda23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:55:44.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92af3750da74559b69bc5a98c5572fe : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1af15910a094ab491092d72ed02c9bc : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eeff18999557414bbac552273ab75349 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cf477f5fdfa4f7292cbe7e1391f940b : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/890394b73469429bbb1015944298dfeb : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1cf877639122445593fc15c4266d892f : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c2592d435f904302b29e00bdcfa2fa39 : SUCCESS in 8m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/27293b39f09e4b92aee5e6e040594980 : SUCCESS in 9m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a1cbfdf170944cc8920150b93ba5bd7c : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4698d7a92c1247f49ff7b4f0fc1b4683 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a6223d559ac24825afe4110dcf81fb9b : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b4c4b299f3d41e4825068831a4ae44f : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca39a34edf0d4f0fbf5b127a375615bb : SUCCESS in 7m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6d8fbbe8f46d4911883d34563e53645d : FAILURE in 12m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4f7e0699b5ce43da82ab6dd4993c1b01 : SUCCESS in 27m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/57e5885116504c8885ae538770a8a6b8 : SUCCESS in 13m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d5e4f0b3e2ab4a19a0d0a60fa561ddbc : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3fcd27ddb8294f4cb6ef9dc5edafa823 : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2bac9a464bc440c48041b8da5d681cae : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c8064eb4cae40b489f9e40c03600dc7 : SUCCESS in 7m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1f9bce27c3a54c779aaddc416366b941 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f66f0ea6cc941c0b4f22fdbb26bbd30 : SUCCESS in 57m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2fd0fe2d7484227bf890ada73447c96 : SUCCESS in 1h 24m 18s","accounts_in_message":[],"_revision_number":4},{"id":"cd4b7c1d37fec2bd8e30c0c111e7821eb9d29284","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-08-07 13:01:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"832bc53e2585daf9452cc2f4db6089505c5b7343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 14:20:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4e9707f8764030834b884bad1d77d1 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791c6e2f92d546108638699dccbb54d2 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe41de21cb4149ef8f8b3b3d882bc46b : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9920138c1e8143a58bbabd24c6d1b7f6 : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c03d67a2ece94188801ddd62a1698a59 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a918b974db504f2ca408e8f7f42306e8 : SUCCESS in 5m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/affbea4a39d94e69bdf872012cf35433 : SUCCESS in 9m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/45fa2221fce04a8295f1abc233bbade9 : SUCCESS in 10m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2c225209463e497e9ec50430a1d9fa73 : SUCCESS in 10m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a129ad3546c4437dae1406d27ae0f2bc : SUCCESS in 6m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6c7fce8bb9e54bd3a789a23b76f339e8 : SUCCESS in 13m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39b9531fb58a426caaf6b983af3edf8d : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9359012477c841ce95552e9fdda03a63 : SUCCESS in 7m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/881b7d2afc2847e3889692096692038f : SUCCESS in 14m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/61196af8ad96441586292261b0998201 : SUCCESS in 39m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7592b740c11f47458d2946cf396b8204 : SUCCESS in 15m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/14786d5f694c46d0ab75756758e0ecd2 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/13c479f97b44422297f559f98300e78c : SUCCESS in 20m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8ff5b144981440a8ab96ddc419c81eae : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fb6f5be2947e45978e5ee0d8a4706485 : SUCCESS in 8m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/54466afb8af54cb2ab02a6e829a1dcee : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c17be08afd94fd8af02a83098b422d4 : SUCCESS in 55m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2283f6d00c774572985d2b9f6238dc98 : SUCCESS in 1h 15m 47s","accounts_in_message":[],"_revision_number":4},{"id":"9bbc8e93caaa4a276211398eae3509e8d98fa142","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 14:23:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/93a4234fa6be499baba71f70cb56ccbd : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c45596d375888a32ee7b020f9c7ea55c9e1f26d6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-07 16:19:14.000000000","message":"Patch Set 4: Code-Review+2\n\nAssuming the depends-on patch will pass most recent recheck and will be able to be merged soon. Once that\u0027s in place, this looks like it should be good to go.","accounts_in_message":[],"_revision_number":4},{"id":"381f86041bf44dec71b80f811b5861115c6cfb7c","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-08-10 13:57:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"cb3b98b16dbe038f5ea7a2644cb903df7e3d8264","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-11 16:40:15.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"dd1f1cc1efd7a0bd7d525329a35217a8323ebb18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 16:40:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"31e99393292b31ff8347737b5e53946a322e1fbf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-11 16:42:20.000000000","message":"Patch Set 4: Code-Review+2\n\nlooks like sean beat me to it","accounts_in_message":[],"_revision_number":4},{"id":"4e84a1edbeb7124403c7eaffc1f2cce7d3858b51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:12:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"194f3bb0b4f1244699d6a1aa746e55377a66e5cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:12:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477c41aa816f4a9090b274a0f8f464f7 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fb7d72633ff40b4a31091983e777847 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74d0b1cdae1343a78107bc4532278d1f : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b4cd3024e4f4a678e7dbf175b9c4af6 : SUCCESS in 5m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7e5b351834f447f1a9bb15739c9cf99b : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9542de36112f4c03a3c2a7df1d09980a : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/24c4b106a70e488eb59c08c9859dec29 : SUCCESS in 9m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/41d64eb6b1e54c608516e24ac875185c : SUCCESS in 11m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/007d0652f4dd48ebba5568e303772cf9 : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d7ca7df908d244aa8ff79a0bc3e7d5aa : SUCCESS in 6m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cd60517e0de0466b839ebd71af248f20 : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc1da936db6a4ca0a996967b07784a4c : SUCCESS in 14m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/508b7137154240979cf568576186429e : SUCCESS in 12m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0f1209e1d6284b57b2be46a0e6eb0f2b : SUCCESS in 34m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9b5267bf9e63443ba07083ae3f5cea43 : SUCCESS in 31m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/74db0693bf27460b8be40827085af6ad : SUCCESS in 14m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b9124714e0a4e44b95007cf101d6c34 : SUCCESS in 5m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53e1aef3645e4cd786c009ca94bcc89e : SUCCESS in 16m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/38435c3ed1e046d6816515dfba7c76e4 : SUCCESS in 12m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17e09b815d784caa800837ca04cba89a : SUCCESS in 7m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7f3a9d1ccb7f4c2c9f1ba2430f5557a8 : SUCCESS in 12m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/030e1c483f80459ebb9ad8a210f7362a : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fe73d1cbed54a48bfe69fcf443c2925 : SUCCESS in 58m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/846c0a06ff2d41b090d72051c8d41ad7 : SUCCESS in 1h 26m 05s","accounts_in_message":[],"_revision_number":4},{"id":"83610c8c7818ccb241ba5b6ce9114a48bbf61551","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-11 18:15:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656bf80687d047edbdc7c177a2e874d9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"15f1946d57b829b763390decd7bc92f8005dc493","revisions":{"765e72bd9666545eec2c691d6c053017f65d69db":{"kind":"REWORK","_number":1,"created":"2020-07-19 06:22:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/741837/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/741837/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/741837/1"}}},"commit":{"parents":[{"commit":"237770babb13510eb5aad4813667a0265205b620","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237770babb13510eb5aad4813667a0265205b620"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-19 06:22:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-19 06:22:44.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I21039783ffcf0bb0257ecbaeeffecea3fca579cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/765e72bd9666545eec2c691d6c053017f65d69db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/765e72bd9666545eec2c691d6c053017f65d69db"}]},"branch":"refs/heads/master"},"ac1f42d3d199ba48da285b203ea62c681fab41c6":{"kind":"REWORK","_number":2,"created":"2020-07-19 06:32:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/37/741837/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/741837/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/741837/2"}}},"commit":{"parents":[{"commit":"237770babb13510eb5aad4813667a0265205b620","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237770babb13510eb5aad4813667a0265205b620"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-19 06:22:44.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-19 06:32:06.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I21039783ffcf0bb0257ecbaeeffecea3fca579cc\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac1f42d3d199ba48da285b203ea62c681fab41c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac1f42d3d199ba48da285b203ea62c681fab41c6"}]},"branch":"refs/heads/master"},"35db6d2a490939d624c466be6fffde3314bb5306":{"kind":"REWORK","_number":3,"created":"2020-08-05 22:32:42.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/37/741837/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/741837/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/741837/3"}}},"commit":{"parents":[{"commit":"a44dfcc12233e940d7d2b9b0f853b49e83d7579b","subject":"update constraint for keystoneauth1 to new release 4.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a44dfcc12233e940d7d2b9b0f853b49e83d7579b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-19 06:22:44.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-05 22:31:59.000000000","tz":-300},"subject":"Update PyMySQL upper constraint to 0.10.0","message":"Update PyMySQL upper constraint to 0.10.0\n\nDepends-on: https://review.opendev.org/744998\n\nChange-Id: I21039783ffcf0bb0257ecbaeeffecea3fca579cc\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35db6d2a490939d624c466be6fffde3314bb5306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/35db6d2a490939d624c466be6fffde3314bb5306"}]},"branch":"refs/heads/master"},"15f1946d57b829b763390decd7bc92f8005dc493":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-08-06 16:52:17.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/37/741837/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/741837/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/741837/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/741837/4"}}},"commit":{"parents":[{"commit":"a44dfcc12233e940d7d2b9b0f853b49e83d7579b","subject":"update constraint for keystoneauth1 to new release 4.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a44dfcc12233e940d7d2b9b0f853b49e83d7579b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-19 06:22:44.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-06 16:51:56.000000000","tz":-300},"subject":"Update PyMySQL upper constraint to 0.10.0","message":"Update PyMySQL upper constraint to 0.10.0\n\nDepends-on: https://review.opendev.org/744940\n\nChange-Id: I21039783ffcf0bb0257ecbaeeffecea3fca579cc\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15f1946d57b829b763390decd7bc92f8005dc493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/15f1946d57b829b763390decd7bc92f8005dc493"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
