)]}'
{"id":"openstack%2Frequirements~741898","triplet_id":"openstack%2Frequirements~stable%2Fstein~Icfa6029688063b7dfed5d08bd009eb5a8106182b","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Icfa6029688063b7dfed5d08bd009eb5a8106182b","subject":"update constraint for castellan to new release 1.2.3","status":"MERGED","created":"2020-07-20 09:38:55.000000000","updated":"2020-07-22 16:35:10.000000000","submitted":"2020-07-22 16:35:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741898-1595435710281-b2062de1","meta_rev_id":"52a97ec43f2d837f87b48000711ab7488581d370","_number":741898,"virtual_id_number":741898,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 16:35:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-20 11:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-20 11:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-20 11:55:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 11:18:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-20 11:55:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-22 16:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb31bf96a352759b8126c748af748937ecf3857a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-20 09:38:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2146c67f49b6a48d1517bca64f04118b4c984ca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 10:57:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778a5de0249942a6a7b9965ab11ee316 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5543cdbd933548e3aa220d1478cb3f77 : SUCCESS in 5m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f57340644cee4fc787696ea0e6c0953d : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03974af7e08e4b74b6ec7f5ce6beea85 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec98e997f9a046358aeb76b8e97a66c6 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7344a3df9db48febdd3950caa1fa8d9 : SUCCESS in 4m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/9310be5ba0664b459d215a7f54f9bcc0 : FAILURE in 5m 04s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/5232f82badee478d95fbffc85735927c : SUCCESS in 10m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb45079b10e54aaeb7e390a413d4ce9b : SUCCESS in 10m 46s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3da2fe180cf2466589111cfb7a505b56 : SUCCESS in 10m 10s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1b37904133a5483d952158a0bcb14085 : SUCCESS in 11m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0426a74283ad4cd784febe9639b95ccb : FAILURE in 4m 23s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/dc5cc98193634c63be0e649e01dcae5e : SUCCESS in 6m 23s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a50617cefe5b4d46b1b68c807530e329 : SUCCESS in 12m 47s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/2cd40a5af8da46459738e1e69d28f5d6 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0c82292e31441f9a281242adf8bc8d3 : SUCCESS in 8m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6da76cff071c4facbedced37e139497e : SUCCESS in 21m 17s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/108a08079ccf4fb8af4dac22ac83e17a : SUCCESS in 28m 30s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/052b03284318489f91f36d43f6aee16e : SUCCESS in 21m 44s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5c4cecf4a0714a74a64ce5d5ca8dbcf6 : SUCCESS in 22m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b7e9743118514adcba3a06ea0e7489a1 : FAILURE in 5m 15s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f624d1677f7e4cb4a87a24297dd068b4 : SUCCESS in 16m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/035dbcc60dff49f489e8c96a7bed5c85 : SUCCESS in 18m 18s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a81fe4c2623b4f1e855540bdb7574326 : SUCCESS in 8m 59s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/fd00c34da769417fafa29f7f40fe602b : SUCCESS in 9m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bc6e027a10748108e7b14a8d6f219f2 : SUCCESS in 1h 09m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/be0038005b2d4e83a7d7f78d09e36f7d : SUCCESS in 1h 15m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c846df0a0b6b465ba45534dbc4533bc3 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d5ab8f5d84c78bbc7008813d9ca7a4d6434cf8d1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-20 11:18:24.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"09b9ef662e5b9864dff7cf95bdedf344eda07afc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-20 11:55:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6dca685bf65a7473b776c5dc885d6b3740671a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:48:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a07c8b65f1149ab9d82013262d86a8c : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/575594b4f8d44e608d87457659c8072b : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c18c7f85c5a544418b2803c88ac87c44 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb6fa87dd94746e2a035d61f3f27750d : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee0f8dc14d8a42b7ba1cb63a00b7c73c : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a1e08ff6cb84ddd8aec4dca3bd0fc73 : SUCCESS in 5m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cdab60b7fff14f8393504fdc1562cc8d : SUCCESS in 9m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d5802e48def048c78a01f406de681624 : SUCCESS in 11m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/737a7b6e5b2e44e296ac28b0e553c434 : SUCCESS in 8m 44s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/bb775212ea794bf48fbf102d5b4a10fd : SUCCESS in 11m 12s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/6e7c37f9921c4393885f57b9bb4bfcfa : SUCCESS in 10m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/01985d13bd3d4119b6717c243211d09a : SUCCESS in 6m 15s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/dc3fe43dca29494d8b4abfcb43ce01aa : SUCCESS in 7m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0f1f1f2074094e1780daf722c5558b82 : SUCCESS in 9m 49s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/dfba19a16b864b189207e74f3af30ca6 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5d924902c6e44e890c2664ef5fde820 : SUCCESS in 6m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8b2b82ba47ff4024a9299c28541b9992 : SUCCESS in 28m 02s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/c8c857e51a184f0d8f87a7dffc0ee13d : SUCCESS in 36m 24s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f4b6466e93e946d69505f8d212004f7e : SUCCESS in 28m 14s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e2de6e2784fa45ce9450c1b1e15f13b6 : SUCCESS in 21m 24s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/8a5261844d594e828a8504a31dc69137 : SUCCESS in 21m 52s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/25879e49eda743ae836907050f2067c0 : SUCCESS in 15m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/154a538085274528a6aeff3e37758453 : SUCCESS in 17m 04s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ff7cc6d28fa34518832a1074c5b822d3 : SUCCESS in 8m 26s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/167cdd5fc1d04a23957f605a14755eae : SUCCESS in 10m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d836975687ef4a19ad4478b8433dc83c : SUCCESS in 1h 09m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/23f375a1ea87472aab1f1833b7f6fbcf : SUCCESS in 1h 23m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fe361d8777f548309fa9e7986d8fa87d : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":1},{"id":"58e185a2656fba42a69e1225a2b8a58035849ec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:48:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"971ad455791dc24ad855032f0b6e68598f817c99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 15:11:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001c38d7f3604987a24288071910cd7e : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb86b46a3f8449abb5334774d17b22e6 : SUCCESS in 3m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1fc6d00f95044da38f22985a369bca37 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21463000e5a04727aba49a6d9a136f00 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe4a8d0cf5f04c89b3f37da168daaf38 : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/de371ae1838c425a8a172292f10c59a1 : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/44a69cef50fe48e29a02d9a751c0ae83 : SUCCESS in 7m 39s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/b9bcc3e85e9041d794ade73bc9f75e9e : SUCCESS in 9m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9d665bdfccce4a058e9cd7b9fef7f14e : SUCCESS in 10m 24s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/1cb786fafe434c71bcdc688442f2e268 : SUCCESS in 12m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/0eaa2781a34b4ec99504cac835160641 : SUCCESS in 6m 57s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/a1347ebea9ad4808a498024088869249 : SUCCESS in 6m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6381059288f64342a8d29adb48b88114 : SUCCESS in 10m 43s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a063996110b140b3b6dbfdffc700cb1d : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/134d62d0169d43b6b4a8a69e318f3985 : SUCCESS in 9m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/276d2d62fd274901bcbab45f7cb4b222 : SUCCESS in 26m 16s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/d5a95082784848de8e429ae91ea3f60e : SUCCESS in 32m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6bfc81c94dcf41679618d52063b45d87 : SUCCESS in 23m 17s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7844d56a2bec49d2b6106fecab461e86 : SUCCESS in 23m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2697bb8efc5a4fc8a3f34f3cbf8e8786 : SUCCESS in 12m 50s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3e063949cd7b4373837f73065429394d : SUCCESS in 13m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e3ef59ca9a14820becc215e305b799b : SUCCESS in 22m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5cd931f98c6545778592c055b03fdc8f : SUCCESS in 7m 54s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/62642207f58b4c09b6517a376eb927be : SUCCESS in 9m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/58d3cb68ad5043049be684945fb8f94c : SUCCESS in 1h 12m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/97113227d72f4cf18484495a2915adb4 : TIMED_OUT in 2h 13m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1a943683e5a84550bb718b078add1d46 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7044bee0351baaa5020cc1a1a1b785322c4e1932","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-20 15:12:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"931d38e0a38db95b5771f3e095cce95d0568addb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:28:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32045a9b1ae445bda9c0bd30124c31d4 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/65d0dcf4eb2d4e09bd864527e0b858a1 : SUCCESS in 7m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d9cdbf8aed014a6fae476a9f656735f0 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab9d20cfd9d14cfab939c8a54eadaf9e : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4eb775da50246b5b7ddd596fe99db25 : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a9764507498c45378bfd6ea2a179ab95 : SUCCESS in 5m 20s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/783f0a11373f4c7d936c17b0c0ed0407 : SUCCESS in 10m 17s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/64c9db1635d14f2592d04b520294ba37 : SUCCESS in 10m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0f57c1011a264e408fad6b43a54afb65 : SUCCESS in 10m 52s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a528bf5a7b8344c6be55ada91d05436d : SUCCESS in 8m 31s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/5d752be9ea04400f8faf923cbf5e1a96 : SUCCESS in 8m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c6529e94011d4c4ba9426b66ecc34af2 : SUCCESS in 6m 29s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b79d21adc38e4916af7fc4259fceb245 : SUCCESS in 6m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/23a13e5d2a20424caa2c77184c7a08e9 : SUCCESS in 10m 59s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/fcf9915895b44b6887114090793ffb1f : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53852a2f6c1d45279836f27382be025f : SUCCESS in 7m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e4183bc9f24b45e595d7d9f32c88fa98 : SUCCESS in 20m 54s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/8d84b5b44f314325ad24e99c18469baa : SUCCESS in 30m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/15d2c6eb92fe45a9b4f85369eaa83d56 : SUCCESS in 22m 55s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/b746913c7e97476cbf8dbad00bec79a1 : SUCCESS in 28m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/83aa12fc153047da9d25409e32c50775 : SUCCESS in 13m 55s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/21e13a867c324bacbf16e118d8e808b6 : SUCCESS in 16m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a9ca0fe3e50149ea865a5c926aa0849b : SUCCESS in 20m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/34c504dab4d34efe8c67bc7eeae164b2 : SUCCESS in 11m 52s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/c0066adbe3e94e3cad13cda9fab6b670 : SUCCESS in 10m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4323022abf50480d8f7a247ac217dacc : SUCCESS in 1h 01m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d829e6dc12e24cb591ed05920d1813dc : SUCCESS in 1h 02m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b65dd30a1a3b416aa5074b1702a2c887 : SUCCESS in 58m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e1d517df852eda3ae003bd18da204045da507c0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:29:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c55b0bdfda2eb0d8b815e7fb6d3ab8ca38295dfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 19:57:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74ed6c73fa83483581e71344b1cbcc38 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8cc4f1aa082f4697ab16d39f879a9439 : SUCCESS in 5m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/52d05fdd3fbf4881a860fe2cdb024ca7 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6478265053dd40c2bfb2c8f45bcb24e3 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2b0571e8298a4b82a9fcd92fbc111713 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fdc8fcfe44174cc88fa92e90129b5ef1 : SUCCESS in 4m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2b712dba369c4876a37f328698291668 : SUCCESS in 8m 08s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/da6ef2a107e146f8b394de68836393a7 : SUCCESS in 9m 52s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a9b18909b2c24092a5ef0b83a75622dc : SUCCESS in 10m 02s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/918275f476cf4fba8195f82a9631cdb6 : SUCCESS in 11m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/41db21dbced146d9965172da2a76ef93 : SUCCESS in 5m 28s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/77d978afd2a24860a0cc482c8463faa2 : SUCCESS in 5m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/67cd7a43e7f44beeb870ab40496dbc9e : SUCCESS in 11m 54s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/55a2eb0e9e224c1198937d2bcac808b8 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad67b41c47da45ad8036abe5154fb105 : SUCCESS in 6m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c82d7e591ba74cef97c610a8702dee7b : SUCCESS in 23m 39s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/2f6e9f394cfa4964a124b1e75dec2349 : SUCCESS in 38m 23s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4de132303529459a8900ec151ce62523 : SUCCESS in 22m 00s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4c07ac0e446943a5a6b6da0fd9a1c01b : SUCCESS in 27m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a48a36bc39724b6081407aacc687e622 : SUCCESS in 13m 51s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/e076f19b625647989d4cbb52fc45d17c : SUCCESS in 15m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a0d69861a064b958eb6ef9502517f9b : SUCCESS in 20m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/921a22f3845040fcaedf0372d162cf73 : SUCCESS in 10m 24s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/9e7a891ee26546a4bf9d8770d5c6601e : SUCCESS in 10m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/da594159b21c44b68ae7e0e7bfdb8b30 : FAILURE in 1h 00m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f4b6ce85897e4147b26dea54f92ee938 : SUCCESS in 1h 32m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e448b823b4114981ba3c79a6fe925e00 : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6b8113ad1624966224cdd6cbbb0ba2ff0e85d171","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-22 11:43:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"df6905420ef54261eaeacd84efb89a5820041c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:21:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0baf2843ea7448cb0b2460977ef7028 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/136d078eb6b841c5935f9667583cb642 : SUCCESS in 5m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/48b7f2f9c8254eaaa70b4c0473c17e9f : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ef70de48a2849759603cd35e1a2413a : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/859b3a854a4149b7bec32af522fa322b : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9c34755b6e84b7dbf04cc971c8c7e26 : SUCCESS in 5m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8c8479008148458aa99b5b7cc7714fd4 : SUCCESS in 10m 14s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/5f726ab7302e4424b81402d3004538f2 : SUCCESS in 9m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/293646c0ad0549fdb4fc22df945fc126 : SUCCESS in 10m 39s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d58947f453ac4513a1b792fdefc4c2ba : SUCCESS in 9m 22s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/5248e203a37b4c839862650213f7c14c : SUCCESS in 9m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4723bf8c604448c2bf3d5d01040ed1fc : SUCCESS in 4m 54s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/6b7c79892d6a42aabfabfd95c76291fe : SUCCESS in 7m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/64cb70febe484da685cbcc4c2b4cfef0 : SUCCESS in 11m 57s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ca27e7c8041a40d085bbfc5b991e1da4 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/891e6131add940fda1441cd7b3031957 : SUCCESS in 7m 51s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7e82ef7cbaa6404cb11106c4c9f22b10 : SUCCESS in 22m 23s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/13c7b4c685474bbfac35a03e2ca9bd18 : SUCCESS in 34m 11s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d95fd312581347799cde8925e78263a9 : SUCCESS in 26m 48s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e089fd1530dd43449d70613ad808c8b9 : SUCCESS in 26m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cae0b0a31f654b0ebca43aa0025e7845 : SUCCESS in 17m 02s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/b6392c765f6d4ebfafc14c48af852d20 : SUCCESS in 26m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e6d0d94e3a84b098c3715c82e7c3edf : SUCCESS in 21m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/664700279f504c7292e63dcb2ce20deb : SUCCESS in 14m 03s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e9e3516494834e0c98e1c571a7718815 : SUCCESS in 11m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8c4402377e424fedbc940cea6dbe56ba : SUCCESS in 1h 04m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e58f7afbcdc242e0b6cf2bbd0670c69c : SUCCESS in 1h 29m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/53f9b43f5f994256970c546c2972dc83 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1a7dc9a498b9ec4021d716db6e89986fdb892846","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 13:21:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4c5c24be30e364e2c55840506e80340500ed967","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 16:35:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c195a40f654942f8a2f2ab3ec3d9e62c : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86a7b54de66e462a940175f6a930cb46 : SUCCESS in 7m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/87fb62306fbf4ac688875ceee04da1a0 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf257186928f4872a1a3c671f6da2ab0 : SUCCESS in 9m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd74cccb05a848e0b3c0fa6d3235f2b3 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9fd41a68c752485687cb8a640fabbbda : SUCCESS in 4m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/702774588cb74d2d99819a86cccab23d : SUCCESS in 8m 17s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/d3db5a890a8b4cb08158d7fe32293183 : SUCCESS in 9m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7bd219b124a740cab9bdcf55e76b57ae : SUCCESS in 10m 46s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/7332d86971244b989e587ae6a98a2211 : SUCCESS in 10m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/60c7f1c9b3194a97babfc08d6da67d37 : SUCCESS in 7m 06s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/174ee166b24843b38717c07efd31df8a : SUCCESS in 6m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/dcb073c8c08e41b7b17396d5d6477587 : SUCCESS in 11m 31s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/98307ae5b3114958977310710979add2 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/be3faf6aa0d14a70b5ee4932187dff15 : SUCCESS in 7m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e5a6c594596c4aae9cf8de05f7a55a3d : SUCCESS in 23m 36s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/3ed904fc229b434ba7eab78f0bfb0587 : SUCCESS in 32m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3e87f598e8514acf80e2ace25c0caa03 : SUCCESS in 19m 59s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/96b0699c485a46428b8017869c5eafab : SUCCESS in 24m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d3b44ec4eeac4216920e4004ca36baf9 : SUCCESS in 13m 21s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d9a6fb6693f84bb2b415f4e266b88e15 : SUCCESS in 13m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73fd646d530540a78fa50b09154d5996 : SUCCESS in 19m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5f0263b93750480a98347b5325a56d58 : SUCCESS in 9m 31s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4dd461cb045b447db6bc3beba0bff032 : SUCCESS in 9m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb8f9f3c160f4cfe9143a5b77b06b388 : SUCCESS in 1h 13m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3606e296dad9414fab1268200b621de5 : SUCCESS in 1h 37m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8564f332c5884f338dd4935960d109bd : SUCCESS in 2h 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"77e8bc1ef7ca7ac7fe3dcc2afcdb6d0e167f5dc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 16:35:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2080357ed432c3b82441fb723b251cca79b96db","revisions":{"d2080357ed432c3b82441fb723b251cca79b96db":{"kind":"REWORK","_number":1,"created":"2020-07-20 09:38:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/741898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/741898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/741898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/741898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/741898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/741898/1"}}},"commit":{"parents":[{"commit":"357d6f5ecf63590cd6bce7d796c7b70b2571885b","subject":"Merge \"Revert \"Make greande jobs n-v for EM and oldest stable\"\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/357d6f5ecf63590cd6bce7d796c7b70b2571885b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-20 09:38:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-20 09:38:53.000000000","tz":0},"subject":"update constraint for castellan to new release 1.2.3","message":"update constraint for castellan to new release 1.2.3\n\nChange-Id: Icfa6029688063b7dfed5d08bd009eb5a8106182b\nmeta:version: 1.2.3\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Ifdd08c9f3457470c6fc0a142c10982bcbfb0450e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2080357ed432c3b82441fb723b251cca79b96db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d2080357ed432c3b82441fb723b251cca79b96db"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
