)]}'
{"id":"openstack%2Frequirements~741900","triplet_id":"openstack%2Frequirements~stable%2Fstein~Ic7019ec0b337886d7827bb6a7750d006477794f2","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Ic7019ec0b337886d7827bb6a7750d006477794f2","subject":"update constraint for oslo.messaging to new release 9.5.2","status":"MERGED","created":"2020-07-20 09:39:14.000000000","updated":"2020-07-20 21:52:15.000000000","submitted":"2020-07-20 21:52:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741900-1595281935290-41dc37f0","meta_rev_id":"05b0ace5a5e1e31bf54cd9d609ed7dcf89aab22a","_number":741900,"virtual_id_number":741900,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 21:52:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-20 11:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-20 11:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-20 11:55:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 11:18:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-20 11:55:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-20 21:52:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec2dda1a01aaf5cd7b1535a1f3501a3cbe0cde9b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-20 09:39:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"545f035799af11b182e1a81566d209913f8f728e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 11:12:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a83362582a19462abbc219fa057d6a5b : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eba784a4cd9349879b63f7406df29a8d : SUCCESS in 4m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a605f560d2564aafaa2b228bb765ec9c : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5d6d2fca4074e75968b07b876ddbe47 : SUCCESS in 5m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe0b2b6ff5d945b286e249fc59d2e036 : SUCCESS in 4m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d92167160ed6449c9bd8109d6f8b9b4f : SUCCESS in 5m 52s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/63aab689a14b42e19a11df66215c162d : SUCCESS in 7m 26s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f60536cc4b204d37ac6deb68e1e78b0f : SUCCESS in 7m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a3abfd5ad3b440be950d41835246cd70 : SUCCESS in 7m 52s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/92de4a45415e46bc9e298bbf13c55780 : SUCCESS in 11m 19s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/c204efc1c0ec4d95bbf9f19044ccc142 : SUCCESS in 11m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8b2c147928f1432eb0cc9e607d2d8594 : SUCCESS in 7m 10s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/7eae9eefb7c042c89df8631f6be18151 : SUCCESS in 6m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/43cff2f0be024ce195f873c37b2014a7 : SUCCESS in 10m 46s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/daa8efdfca53414c93ba48959a6e9901 : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1163c412e7a44f4a9c72da7ddba3cf76 : SUCCESS in 6m 31s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/054b7812fd364699af2f5dcbe52c5752 : SUCCESS in 23m 33s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/4c2e5e451f004690a46df4598efee7ce : SUCCESS in 31m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1a09bdd66cb24603999505a33a9eb997 : SUCCESS in 22m 47s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/510155dbd21b41f48c806b6b31497dc8 : TIMED_OUT in 41m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/17c2d6250239471eb7631357061ac394 : TIMED_OUT in 41m 27s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/aff57417a17d40819227b09073da8163 : SUCCESS in 14m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7534441073342d58723f3b1a784208b : SUCCESS in 18m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d05cb4e08e9944c5995cfdd1dfea2110 : SUCCESS in 7m 51s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/9785f261188a422ea28e059378cfb17b : SUCCESS in 9m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/22d1098b2ab248d0ae8105ecba9d5289 : SUCCESS in 1h 09m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/eeb10cc282924405abf233d86cfbae73 : SUCCESS in 1h 26m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c6fe7761e62d4618886454e522dcddfe : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2de547b8abbf749e1626783d206fa54120530840","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-20 11:18:43.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc06fe2722a969720acfa2c1826bdcc998de91f8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-20 11:55:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2196f7fb1389515e8f8c7e55ae3aa6f10e7ba9d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:51:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88921fb8cd8f4805972be47677891060 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73c97c9ed9e3407ab7e73aaf550b4a7a : SUCCESS in 5m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/36c3cf01679a456b83d15866282c0905 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdcbb0f77ede474d97b5a01493e7661f : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c61b608a677471aaf7c06e2bff63d33 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a31aa0a3389641c1aa09c94ee96e95d2 : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/903f727186434eb7a215659289baecd4 : SUCCESS in 8m 42s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/b8b40b8a836f4e599b4606ac4d79b37e : SUCCESS in 10m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4025e40b93314eb7848e52a45c60d8ee : SUCCESS in 8m 48s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/422e3136ef3048bdadcef4719e199595 : SUCCESS in 9m 32s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/dfca413dac6c449392d46fe2791b164f : SUCCESS in 12m 39s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e08c5fc980c54e5ba736ab409ee45c70 : SUCCESS in 6m 48s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/98046ad94ca64cc792023f75ce928571 : SUCCESS in 5m 47s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0d062478846d425ca32e5f402109d369 : SUCCESS in 11m 02s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/378b0cddcf5246bfbedbf7a99c1b511d : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24cb0f8235414792a37a6a1f8133bbac : SUCCESS in 6m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/771253984a7f43e88aa0ce52433a5233 : SUCCESS in 21m 04s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/ecc1d7248bf843ba9156caef0d99f47e : SUCCESS in 29m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6d7abbb772b8407d9c9a89f2376b771a : SUCCESS in 24m 10s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ef451d3630d24c118def365d8963dcbf : SUCCESS in 23m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9b8c7338411d4a7cb638a90a31a1098c : SUCCESS in 13m 27s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/fba75cec08e743a9a05e42876a907df7 : SUCCESS in 15m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/37042a602bd647e682dc62cc280a9331 : SUCCESS in 18m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/853e5076e5674370bfe02ed95c549a31 : SUCCESS in 8m 08s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/6a459b7df106459d81cd007cddf2c616 : SUCCESS in 8m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb6fa418874a4f7d9acd178ef826cf6d : SUCCESS in 1h 09m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/98a2ecc9ecf54287a363fab9af7c86dc : SUCCESS in 1h 27m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/381ff4aca71f449b80469aaf9b806868 : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":1},{"id":"33b150694131fec5d37f455f166ead6d0ef60420","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 12:52:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"411528d21c866bb52afe3c692773623457f75f62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 16:50:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d860f083c3194918844542f169659e91 : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/594523d34d164da0879282aed637987a : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c138e46763a44ac8797c4f3dc775f39 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0be46573f57d40cd94cbfbb9f8bf3b02 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c684b8f4884142f2891174970ea28cb4 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4988eadaa6bf4b1591d6ef7c2bf46d1f : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c36d81975116496480365fd1fbaf5be1 : SUCCESS in 9m 52s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/9c7287da71924ce9876f39e280aae1fd : SUCCESS in 10m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c33b8b7f6a974864a2d99bcc836d6de3 : SUCCESS in 12m 11s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9d93071c07544ac6a841662b829b7112 : SUCCESS in 10m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fb4f5a796eba489881cf8fe457ccbd5d : SUCCESS in 7m 57s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/1c4351c3c08b432da28455ffe84981d1 : SUCCESS in 5m 38s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/8c6a46cbda36483cb63ef9e17d7efc18 : SUCCESS in 9m 32s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/3e329f60bfa446f4b5d87c55662c9924 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a227eb1cd3f549669fe23ef903a0ef3c : SUCCESS in 8m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/810357a60b3d4e7ebf95af56f5a8e10c : SUCCESS in 24m 56s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/f84a9d0c6ed149e19b2e8eb21123c292 : SUCCESS in 29m 52s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e279ab1ad49b4aecb9c1a76b19ec4318 : SUCCESS in 25m 36s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/15222b5dfb70476aab36ebb0991ad5c7 : SUCCESS in 22m 39s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/311e9f5feed24df993c5a7b405da62ed : FAILURE in 12m 29s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/712c260602e84318b1d6e11defa89481 : SUCCESS in 15m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9811069fc7a24a25923a9736fff5161b : SUCCESS in 18m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4c1adbd24ee042deae3d92f2042adfe2 : SUCCESS in 10m 45s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/49e1f50190cb48ac911438e58fa46a80 : SUCCESS in 9m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a613dc714d4b4a35b4081b9ca11ebf80 : SUCCESS in 1h 18m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f8b46a50be49405cbe2767efa855a559 : SUCCESS in 1h 30m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/19d02529b6ba4d829d9cd3d9646be0b3 : SUCCESS in 1h 09m 00s","accounts_in_message":[],"_revision_number":1},{"id":"abfc9d3c816794f547a5498a7c4dcb600fcfaf23","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-20 16:58:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e825b9a2e4c94c6211437fa882e82a9f9f6f0ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 19:27:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dec676aab9b74b018a7056b2f6ca97c5 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe41e5e391de499eabdd1c828f83da93 : SUCCESS in 4m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b3168c7c87441409d0dd90ccd4c0188 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/305efad203084be5ad3898d0ea309162 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9dbb7773445c471f90294881ad91413e : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/353a353154d343d6a53dd78249bdf722 : SUCCESS in 4m 14s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/778a9a7ebb8f44f1a846ad84a595534a : SUCCESS in 7m 51s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/3760ee8e7c074f4593da0970297d95cc : SUCCESS in 9m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fa53ff593bea42a88e56c43f2f0fa32f : SUCCESS in 12m 39s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/70f3fcadbce34e51a2650c222ff77846 : SUCCESS in 12m 19s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/9a782a2ef7594d199d1ca2d5186eb359 : SUCCESS in 10m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cfd9129a2b25423a992dae5667ebb29c : SUCCESS in 7m 40s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/555c2c151cdb40e8ac6f509130b42d8a : SUCCESS in 4m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/33f2392b886049c6bbf38b386238112b : SUCCESS in 10m 59s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1fbc195d96054b5bbfbda9802b701e23 : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e83aeee2393469491f17718fb351d03 : SUCCESS in 6m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0f41dd9ee08d431d870fa12ca994fb79 : SUCCESS in 15m 09s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/3e6913fdf08e491582ef81f828be14af : SUCCESS in 20m 06s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6b7d1758ff1f45bab58102c439eb3bf3 : SUCCESS in 15m 49s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/de4e3d36952f4b99b6fccaf9013bfb25 : FAILURE in 14m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6ea267d9b06b4217abb14997e63a7844 : SUCCESS in 12m 42s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/4ffb57c71f874561aa20a8790869b17f : SUCCESS in 13m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/157836d2c3334f83862eac7bedac549a : SUCCESS in 19m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3e1b4c98fb214434a0cd0a88ccd4d4a6 : SUCCESS in 10m 49s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/dbf3dc349efe4069b0363e52394b89d6 : SUCCESS in 7m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e90e6f45668b4b0b9edbcdfe45077c13 : SUCCESS in 48m 52s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bda3b980ee364cf89cf1390edfdfc264 : SUCCESS in 1h 39m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/468b9854ccc946a3b1b7ca0f79cf210b : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3377a7daf3541435d012d9074c230150ca84ea6c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-20 19:28:42.000000000","message":"Patch Set 1:\n\nrecheck\n\nLet\u0027s see which one breaks next.","accounts_in_message":[],"_revision_number":1},{"id":"f407be8ac8fb3f1930923ac87d1876d4a6604b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 20:48:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4228526b784440c4901c17a528bb676d : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/130a13415c9b4f63a78ff0b20a7a2375 : SUCCESS in 5m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aef1e6173e434f1e9d85ab893d238cc7 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cc1ad65ce6f4b989293d2358a42ae28 : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22f8e32c97f345f79d269df3de62d691 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ada50e4ca6a48af8784943a0c2bd880 : SUCCESS in 4m 12s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6dabee9eeceb43e6b8052f406045a9df : SUCCESS in 9m 51s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e3720852cbe44cbc89687d01298e236f : SUCCESS in 10m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2fcd0180d3944ac685b407e5752b9ce6 : SUCCESS in 9m 06s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1967b87418e04e23b8dd640cea8c46ce : SUCCESS in 11m 57s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b1aae78ab0f04b958cf66d14c512ee5b : SUCCESS in 12m 09s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3c913bcd23084543afd8567879fe4ea3 : SUCCESS in 6m 14s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/5a3c5c89c7c6484d99524c16e204029a : SUCCESS in 5m 12s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/6511b5d5c8ef4432a30c011d49c37dab : SUCCESS in 10m 58s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e0a7b93368c44d6aa4bdc4fb31329bc8 : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/138665a1762a4294bc0b7c2eb5e06962 : SUCCESS in 8m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/87326cdcccd246f2a8fa82f266ab080e : SUCCESS in 23m 40s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/9fa12d1f367a4baba65132f031b6135c : SUCCESS in 27m 04s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e8544bab3cdf49df88e21f8705d27fcf : SUCCESS in 21m 55s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/176e2affd803462f904d2689183675e4 : SUCCESS in 28m 56s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/9bdf51e56ca34cd79ef14b2f4dc30e51 : SUCCESS in 13m 30s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/5ad1e4a355284e4d9f815b4b2ab70ba9 : SUCCESS in 16m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7112e87977e947aba8423d1411807609 : SUCCESS in 20m 30s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b398e38c282544039c92a4a08e484fd4 : SUCCESS in 8m 59s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/3e5d42de1e1946a0a47fe211948118b1 : SUCCESS in 10m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ba495e8cc5549f9bb41be5887c5b625 : SUCCESS in 1h 06m 58s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f5d546c136ab43beb2117c04e4ed99b4 : SUCCESS in 1h 17m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bf4720dedf6849e48ef302565f3ab4c5 : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ec798603791de8652b279ed873c7bebff6d36100","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 20:48:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dbccfa3dadfb0fc110726cdead0316f2d39b8e6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 21:52:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e9583d743b0ce7b6a607de95a15294b07b927cc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 21:52:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8048ee6bc9a413080d2ac297f8bd0f9 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34855bdaf57a468d8178241cd93ae3b7 : SUCCESS in 4m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/117271719234442886966843b0f6617e : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f99082961c6a4a23a0922560b555bf3d : SUCCESS in 3m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de36f13656394491a2844518a2286328 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1dc8db7945c848d4a9c31fdbe577273a : SUCCESS in 3m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a145562197534363a109fe55cab076b3 : SUCCESS in 7m 21s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/eeb37ef37ea349e795c4c60547e50d07 : SUCCESS in 10m 59s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/76db94e21894483aa01fe91cb205d0d2 : SUCCESS in 10m 02s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/5f73c71df4e54bbf932e37dfbdcdd6f7 : SUCCESS in 9m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/6bd0e9011d854b6797af10000d755b8e : SUCCESS in 5m 01s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/44330f2252ab4384bc3250b9e0b542af : SUCCESS in 4m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/062a5f7d3bf741719d689349066b7a2e : SUCCESS in 13m 26s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/25e58c801c7b423f85ea48aea218b19e : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec7a5dbc14cd467a8c23e7fe2cab4044 : SUCCESS in 6m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/057a16a399e84344803d6e70375a7fe0 : SUCCESS in 13m 42s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/31e02cdf6a6946e5a78877a0d0bdefb2 : SUCCESS in 34m 09s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4f5694bab1de4b5bbd879cb877f76f8a : SUCCESS in 21m 14s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/0cc83dc5a0e44701a30250c3a46b945a : SUCCESS in 27m 04s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7a1648d72e204c0d80aac68ef6119a0b : SUCCESS in 13m 09s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/8588982724f94852859cdb27cb9e56f6 : SUCCESS in 13m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b7b7becb63ee43e49649cbf9be39ddf5 : SUCCESS in 17m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3257d81c7dfd4bfb85b1aa1829177c18 : SUCCESS in 7m 01s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/7fbe1547e9e1437eb45f172c8899eaf5 : SUCCESS in 7m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4efd460202824ff59e567d6d6e7d8af2 : SUCCESS in 45m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0a82113eadc44404bd5b787f599a8164 : SUCCESS in 55m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ec246e48ec924cff85c6d0383f0a3103 : SUCCESS in 35m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e4c22f53bc314d483eacb4032b16a32bad59a6f","revisions":{"0e4c22f53bc314d483eacb4032b16a32bad59a6f":{"kind":"REWORK","_number":1,"created":"2020-07-20 09:39:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/741900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/741900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/741900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/741900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/741900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/741900/1"}}},"commit":{"parents":[{"commit":"357d6f5ecf63590cd6bce7d796c7b70b2571885b","subject":"Merge \"Revert \"Make greande jobs n-v for EM and oldest stable\"\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/357d6f5ecf63590cd6bce7d796c7b70b2571885b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-20 09:39:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-20 09:39:10.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 9.5.2","message":"update constraint for oslo.messaging to new release 9.5.2\n\nChange-Id: Ic7019ec0b337886d7827bb6a7750d006477794f2\nmeta:version: 9.5.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Ifdd08c9f3457470c6fc0a142c10982bcbfb0450e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e4c22f53bc314d483eacb4032b16a32bad59a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e4c22f53bc314d483eacb4032b16a32bad59a6f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
