)]}'
{"id":"openstack%2Frequirements~741906","triplet_id":"openstack%2Frequirements~stable%2Fstein~I9142b5ffc53c06103234f24236900d63e4458a82","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I9142b5ffc53c06103234f24236900d63e4458a82","subject":"update constraint for horizon to new release 15.3.1","status":"MERGED","created":"2020-07-20 10:01:43.000000000","updated":"2020-07-20 17:04:46.000000000","submitted":"2020-07-20 17:04:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741906-1595264686951-a8fb295d","meta_rev_id":"c3b45712c56a483d9661d6564cc9e5b3c71f5fd4","_number":741906,"virtual_id_number":741906,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 17:04:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-20 11:55:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-20 11:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-07-20 11:55:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 11:16:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-20 11:55:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-20 17:04:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e890839456c33413172cd247c7d6e645042319cd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-20 10:01:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2658e2debf6473847d4011f9c30c721b26d2cbd4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-20 11:16:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e758da62d67a054a0c3a38e1d3d2ae4b217656b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 11:30:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f15d6d31564ec781d6e5ac31ff24b0 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/37614279ae0447a4abfa85b6f34b86c3 : SUCCESS in 4m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ab0edc4ebfa40c6b57295ad9b1827d2 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2c584a5002b4faca46501762d1c06c9 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4d67bd2f6705415699682e7f6caad183 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/642bbee0ed0445579ee017a9aaa3eb7a : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e84ca42619ec41e386f37cbf4e82f934 : SUCCESS in 6m 31s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/7586e64d561348f980d1316933fcadd7 : SUCCESS in 8m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bfd65042219f480d83304f8f135ae188 : SUCCESS in 9m 10s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/111eaa171856458ba4777e0e217c5fc0 : SUCCESS in 7m 38s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/6fb0a1d5483248ff9dbba44e9a56c2de : SUCCESS in 9m 29s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8ef6dd7756684dbd8a273f1e1d4b411c : SUCCESS in 5m 35s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/f3646af70a2a44749e7ea141044d66f7 : SUCCESS in 4m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/be4fc3a77f43471d8d344217a889be67 : SUCCESS in 11m 53s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a284e928a71a45faa6d3bb6495be0b85 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/59a2586be9764c7ebbf95c63b1bc4c1b : SUCCESS in 7m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ee017cfe40204828a4de8501b1fa7ea4 : SUCCESS in 19m 40s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b34fe83a376844cbaab46f96a022df3a : SUCCESS in 29m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/aa1cceb951d04623863a20b05422eb1b : SUCCESS in 20m 45s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/b3136707055d4491b78ec270733a8d15 : SUCCESS in 22m 40s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f3e45c84f4bb412e9b362a8f05203153 : SUCCESS in 13m 32s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3a3e383555294137b3be29b5d5784919 : SUCCESS in 13m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c5f775fefe44f4d8af921236e3fe4f7 : SUCCESS in 14m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f0b568ec84c34d89be0d2c73131ebee3 : SUCCESS in 13m 03s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/d3d55e6689fb463ca71f7468c921af89 : SUCCESS in 15m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c84b7692811f4cbfae772920d21165e2 : FAILURE in 1h 05m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cf5ad98149274353b409de77f282b136 : SUCCESS in 1h 21m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8722d92d85af44bcb74b7cee5c970856 : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f0a749c4f79ee49796ef3c7b78190b049f79b1ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-20 11:55:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"82857e7bf40cb328245ff31a32c5ff65d492aa59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 13:28:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfebf81edafd4a049e8d2118114a5219 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1f1542d6425433398fa7a1d6236ead2 : SUCCESS in 4m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fecf303881bf4febb6b4eaf158caa806 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08f1e953e7964bc89c25d1afba8f62d5 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/526d7b6b7f8d47bbb51deae43c8cecb9 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/212f51665eed41d3863616df4366981e : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3e82ffe6531341d187e138c4053cd69a : SUCCESS in 6m 56s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/9e689dfcebe84cbcbdb06cbb72ad3466 : SUCCESS in 9m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/143b281a8e2e4297981d1e23ce5a93ec : SUCCESS in 9m 56s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a39e4f4e0b56476baea53a0f987ea065 : SUCCESS in 10m 49s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/78c40c09d0ac4cc4b3e1238f082d4443 : SUCCESS in 9m 42s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/25cddf3a6b5b403da40c640b11be6f6f : SUCCESS in 5m 15s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/160aef9fc8b34142883d27642a4d94f3 : SUCCESS in 4m 44s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/50d9d3ecb5fc4aac92081a1974505158 : SUCCESS in 11m 17s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/88dfb2875b8b4b96a4ec63a4a2b72dab : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20d05b82114b4cb1be71c70ae5e5c46c : SUCCESS in 7m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8794de86e6f148e8ac12aab779a86b20 : SUCCESS in 14m 29s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/ac825290145142ae89cd128b33bb2dbf : SUCCESS in 24m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d30c45e474d64c05997588ff04dfae01 : SUCCESS in 20m 03s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7cab69995c024d8ebb7de2616ec286fb : SUCCESS in 29m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ce373bd0bd2746f68a82c25acd2b74bc : SUCCESS in 11m 13s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/55ca02d889dc473aa01613e36c7608ee : SUCCESS in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8c2134a4ccd4fb3bcd82fe5e2e33b1c : SUCCESS in 17m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/85efc0b2f7444afc9ee66ddba0faf40f : SUCCESS in 7m 50s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b38cd52eba0248eba58d220c8630e035 : SUCCESS in 6m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/02242d45ee414fdbab73163702eeae98 : SUCCESS in 1h 12m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/39bbd012180f4967bf710d5d7a52274b : SUCCESS in 1h 13m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d559217dac91402ab150ecb386857e4f : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3ba5405cac379864137bbb5e932d977ec36bffee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 13:29:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10914d8ae4facccfda37db846f4dff1232a60dfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 17:04:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b4f05fecd1baeda43946434a9c3ed2fd5da5a653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 17:04:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8747487b230c47d48c76c783282e6751 : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ede590d2528d4419af65bdae72f43f07 : SUCCESS in 7m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c850e67a12542c2b3c4a5d062aaf80a : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68eaf0eedfde4ba5b0fad6b6091be116 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfc90cb0cf8945818d7b3b635f36995f : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b44d91697a2444629f916798d66c7d4b : SUCCESS in 4m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2d9bf364e5d24798b246799610d4a780 : SUCCESS in 8m 57s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/631826d6b13b4cb8bdfc0e6604ea92ea : SUCCESS in 10m 33s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/fe0c351d32014102a648a3fff710b465 : SUCCESS in 10m 08s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/01004ea1d84344c19c1f0779ef5fd717 : SUCCESS in 10m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d88a8c2fdc01440db9ebea83b5d2dd55 : SUCCESS in 6m 34s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/9fc2f79e80644ef88f849701e921bd76 : SUCCESS in 7m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c15a9ebf382a4831aa3ac27051eb06d2 : SUCCESS in 12m 40s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/2a50a0d93b654766aff19db1c38f5a89 : SUCCESS in 13m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81563dd238614f4f84dc1566f8fb48d9 : SUCCESS in 10m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/21ce236aa2824233badc763a3c5d0a10 : SUCCESS in 27m 35s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/e9f9cdcfd3f445bb9b33bae8c6661059 : SUCCESS in 31m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/4dbd5a6f0c95472cb678c8cf4cbf1fd7 : SUCCESS in 15m 06s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/e510642ea8cc4e13a7eea16663b7f2e2 : SUCCESS in 22m 27s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b2785d18554c4ca7b1bc798372ddfacc : SUCCESS in 11m 19s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2666888385724089b1642727f2848bef : SUCCESS in 20m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bdaa79b06b61421290a315439b540301 : SUCCESS in 22m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6d01f146fe6b49e8ba1b97b9e899795a : SUCCESS in 14m 18s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/6b7045426ef744a293d4b6f735c217ec : SUCCESS in 13m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/95df6d7591a04f87a69295480ad3480c : SUCCESS in 1h 05m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ca7686c4f9d941dd96b03c628a1b408c : SUCCESS in 1h 29m 03s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c5462134faf04717859e11e3c55c7bb4 : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a163dd33a64b4bad920565582d3d17c6eca73d70","revisions":{"a163dd33a64b4bad920565582d3d17c6eca73d70":{"kind":"REWORK","_number":1,"created":"2020-07-20 10:01:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/741906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/741906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/741906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/741906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/741906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/741906/1"}}},"commit":{"parents":[{"commit":"357d6f5ecf63590cd6bce7d796c7b70b2571885b","subject":"Merge \"Revert \"Make greande jobs n-v for EM and oldest stable\"\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/357d6f5ecf63590cd6bce7d796c7b70b2571885b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-20 10:01:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-20 10:01:38.000000000","tz":0},"subject":"update constraint for horizon to new release 15.3.1","message":"update constraint for horizon to new release 15.3.1\n\nChange-Id: I9142b5ffc53c06103234f24236900d63e4458a82\nmeta:version: 15.3.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: If013d4d1ecec13d1b6d5c09e5f1888d4c60a91a8\nmeta:release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a163dd33a64b4bad920565582d3d17c6eca73d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a163dd33a64b4bad920565582d3d17c6eca73d70"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
