)]}'
{"id":"openstack%2Frequirements~741986","triplet_id":"openstack%2Frequirements~master~Ic2a38b77b5c217f769f67ed1c45d3aeb90d9cbff","project":"openstack/requirements","branch":"master","topic":"bug/1888237","hashtags":[],"change_id":"Ic2a38b77b5c217f769f67ed1c45d3aeb90d9cbff","subject":"Pin dnspython to 1.16.0 as eventlet fails with 2.0.0","status":"MERGED","created":"2020-07-20 16:30:31.000000000","updated":"2020-07-21 08:29:13.000000000","submitted":"2020-07-21 02:58:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741986-1595300280582-78954b85","meta_rev_id":"0bc914e2c65686d2fe53a37d5b3c1b53b6d63a9a","_number":741986,"virtual_id_number":741986,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2020-07-20 19:12:08.000000000","_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-21 02:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-07-20 18:26:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-07-20 18:36:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-20 19:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-07-20 18:29:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-20 19:53:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"date":"2020-07-20 18:26:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-20 16:32:05.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-07-20 17:26:39.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-07-20 18:26:02.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-07-20 18:26:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-07-20 18:29:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-07-20 18:36:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-20 19:12:08.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-07-20 19:53:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-21 02:58:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42f88c8176cd33841f724c4ba1e5f260b8b1c97a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-07-20 16:30:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"352a60da2b5acda9bb237373c7c4f804545af0ab","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-07-20 16:42:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90ecacfeb510b1eb3da3fdef3e3d8f5d87e789c7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-20 17:08:45.000000000","message":"Patch Set 1: Code-Review+1\n\nWe need this patch in Neutron too: https://bugs.launchpad.net/neutron/+bug/1888258\n\n\"greendns\" is not compatible with this \"dnspython\" version.","accounts_in_message":[],"_revision_number":1},{"id":"54be286b8d3ce3b33535fc66cbe388d7c2033f43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 17:10:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1f2c116da0564a3aa67fb0e037e06be4 : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":1},{"id":"59a0c26f509c05e184460b00c225a3f6872528af","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-20 17:24:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c159893e708e1c4a74df872b78bc2f39c5f248cf","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-07-20 17:26:39.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nNeed this to unblock the nova gate.","accounts_in_message":[],"_revision_number":1},{"id":"db1308c3dec6ab010fe5e7871d3886dec84a80ef","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-20 17:39:49.000000000","message":"Patch Set 1:\n\nBTW, related to this problem in eventlet/dnspython:\n\nhttps://github.com/ralonsoh/eventlet/commit/46fc185c8f92008c65aef2713fc1445bfc5f6fec","accounts_in_message":[],"_revision_number":1},{"id":"7516fa260f904c474e9e58ca83f18aada4484bbf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-20 17:41:47.000000000","message":"Patch Set 1:\n\nnow we just need a new eventlet release :D","accounts_in_message":[],"_revision_number":1},{"id":"98486b8131b2fcf49ed07821a7a2ecd013741d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 18:11:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064698338d0341b496f5d4637fa5573a : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3565b12629e44e1944288b4492aeb7b : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96bffdceabaf49059f9980d87f41441e : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95833806e7e9467d8240eea703368e6f : SUCCESS in 5m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cc22630ad17d47c78eac8b2f7752291b : FAILURE in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6a4bd04a3dc4426b5fe1bd82e07fd97 : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7c07beb01bb9494080a1677e8ae96013 : SUCCESS in 9m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd918d9a894e4e1a896106bed630fc92 : SUCCESS in 9m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/86a7877098f74711a507f8f628f212e1 : SUCCESS in 10m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3da0d9bc75fe4618bde5831da765df0f : SUCCESS in 7m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d94c16d8857749978e41a2e806fc8d58 : SUCCESS in 8m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1aca8ede29ba48789f282adbf31b4e5e : SUCCESS in 6m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f431baa610164c54846a0cadb6b60a71 : SUCCESS in 10m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eb88c69c033f4c68b05f727fb8873f1d : SUCCESS in 10m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/db68ac91a1934bf3b97c1b479997076d : SUCCESS in 25m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/61e9e458dc2249c48ba3db302cb08c89 : SUCCESS in 15m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5c824cb0f84949ac8f0c4e4f597f2d90 : SUCCESS in 18m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b61cb292a7b649e8950474deb7b571ec : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8e6a2ecd6a6419d8c0992fa896519f0 : SUCCESS in 7m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/90efc161da5541f8971c3004b4f1bc4f : SUCCESS in 11m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0320de1be1104b8e8c5bacf6f80ebf64 : SUCCESS in 1h 01m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46fbfe8345c14736afca60c78769afa3 : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":1},{"id":"933f0b9b1faa8cf5310a3a04aa678ed8cbacafed","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-20 18:25:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"200231abb5e60055218e06cef9881754a2281e48","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-20 18:25:27.000000000","message":"Patch Set 1:\n\ntox validate doesn\u0027t like comments in upper-constraints file","accounts_in_message":[],"_revision_number":1},{"id":"5195bf904400f9b509c86fcc15087d5524dfcd5e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-20 18:26:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5b162b581dc3fc36136039e52d6661449ad33370","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-20 18:26:45.000000000","message":"Patch Set 2:\n\n\u003e tox validate doesn\u0027t like comments in upper-constraints file\n\nyes, it doesn\u0027t [1]. I\u0027ve removed it.\n\n[1] https://zuul.opendev.org/t/openstack/build/cc22630ad17d47c78eac8b2f7752291b/log/job-output.txt#681","accounts_in_message":[],"_revision_number":2},{"id":"eec00f655b8dbc59826b8a91eb107e5a983fe295","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-07-20 18:29:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e556030771e4911d2e041c150114357dcc0069a7","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-07-20 18:31:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"426b79e014b51ad52900302ee0366bc836c7c6fd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-20 18:36:38.000000000","message":"Patch Set 2: Code-Review+2\n\nWe could cap it below 2.0, but that has proven a pain to uncap in the past. Being optimistic, let\u0027s hope that there will be a new 2.x release that will actually work and we can just block the version(s) we know are bad.","accounts_in_message":[],"_revision_number":2},{"id":"48ba5a0c166137eb4546f6efeb79119d5d573f60","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-07-20 18:38:48.000000000","message":"Patch Set 2:\n\nby the way im not +1 or -1 on this change becaseu i know my personal view differes from the requiremtn tesms view.\n\ni kind of feel like we should be caping the major version by default and updating them periodically.\n\ni also know the rational for not doing that and following head. that said when an indriect dependcy announces a major version change like this and breaks one of our direct dependincies and we dont assertivly prevent it happening again it feels like we are repeating the same mistake.\n\nthe major downside with capping is we need to activly reasses it and make sure we dont fall too far behind so that has problems too.","accounts_in_message":[],"_revision_number":2},{"id":"7525399ad23e1c1f9464c6c8257e35fbe06e04cd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-20 18:52:16.000000000","message":"Patch Set 2:\n\nIt would be interesting to understand how the cross job for Nova did not catch this.\n\nWe \"reassess\" our limits with every update to upper-constraints. So we shouldn\u0027t really need to cap major versions since we implicitly cap *every* version. The testing usually is really good at catching these kinds of issues, but apparently in this case, none of the test cases run must hit this code path.","accounts_in_message":[],"_revision_number":2},{"id":"6b39edd481a92a0ffbe5e307ebf2d79954afeb1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 19:09:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/68f191f96b6a4a92a6000a822a39350a : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":2},{"id":"884639f2390b6eb2546e22917465a5ad861d9282","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-07-20 19:12:08.000000000","message":"Patch Set 2:\n\nthis only breaks in the nova case if the test_novnc tempest test is run and you have enable tls and have specically also enabled tls for no vnc.\n\nso as gibi pointed out in the bug you need to set\n\nUSE_SSL\u003dTrue\nENABLED_SERVICES+\u003d,tls-proxy\nNOVA_CONSOLE_PROXY_COMPUTE_TLS\u003dTrue\n\nin the local.conf and the only job that has that combination in nova is nova-next\n\nso we could have caught it if we had that config in the cross-job\n\nNOVA_CONSOLE_PROXY_COMPUTE_TLS\u003dTrue is proably whats missing as i think the tls proxy is enabled by default now?","accounts_in_message":[],"_revision_number":2},{"id":"b921e9f8c25a42d11c755a09e3e5e9e60e0571d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-20 19:53:07.000000000","message":"Patch Set 2:\n\nWhat is more likely to happen here is IF eventlet has a release with the fix, we\u0027d revert this while updating eventlet, then mask the known bad versions of eventlet.  The bad part about that is that we\u0027d basically be raising the minimum version of eventlet.","accounts_in_message":[],"_revision_number":2},{"id":"235743547075c03584b8e2ae52cb1e70d6bec4ae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-20 19:53:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"116ed9c1169ef655f3283964f60fa80d6b961a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 19:57:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7fa79ac5164887858fcec59ac52031 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b811b764c4422e96fe8fb3ce20cf06 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e648863fe4d48838cc24f70c9efb189 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f224246102fc4cb08c0f18a8adb36223 : SUCCESS in 4m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f6ea97ad5614ffda98d87e06f92c75e : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b608c51e0b7f42b3a793f269a063daf5 : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8aeeae7fbea44d729642d4c157ef8cb7 : SUCCESS in 12m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d0f250aa175649a89514fe9bfcc5bc5a : SUCCESS in 10m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/330b79a28bcf4e119438ca824e34b8e8 : SUCCESS in 10m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3cfd233b01584118897f37ca529bbb63 : SUCCESS in 5m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/61499cae50ae406aa504560a0bbd82b1 : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d5df8923a8b458a80554ce369f7b47d : SUCCESS in 7m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6bfdfb0428c24043a1af5d80c7b4a81c : SUCCESS in 7m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c555136925554b0abbe30969a6e21beb : SUCCESS in 33m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d878623c59f845fd87d8963f82a34191 : SUCCESS in 25m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b4fe6207d9c84299b5ca4a0874946aaf : SUCCESS in 9m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0878057e61404cddbffd41dd0569cec3 : SUCCESS in 17m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fc92a0f200f04b2cb97cbf43ef3bc32e : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31c9605c8a3e41c9b0c45ff0f26efd67 : SUCCESS in 6m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fa3fcd206ca3471f9d022f49547fc082 : FAILURE in 15m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3bbcac104140484b9fce6a565c1687d9 : SUCCESS in 55m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ee53d151a6694cf6a1008a5dc49cdd87 : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":2},{"id":"c41b82f1a1651f1e5c20f5bc26abac3af02631fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-20 20:00:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cdef4d0e907c56d09540e5042ca568b03d92b937","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 20:42:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4a61c85ada2945e29de6ba40d5196773 : SUCCESS in 37m 17s","accounts_in_message":[],"_revision_number":2},{"id":"82b62a3226db5fedf7dbe3ba40a9c2fd038864c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 23:10:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc19f01a8b7845559d2fceb5dccdf26c : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b878b04f3947c1a51d2662fe80c444 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b438426b5824fdc9d0deb552ebbeb1d : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fd69c5ad859466cb221ceb461fe0ec9 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28c74667b1104f808b503cb1b164481e : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6ddc9fb542f447d792e513b2f20b9026 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/55c0e4531994427c836d98853f1e53d0 : SUCCESS in 10m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/86f916bd9b6147298a278aa67108ccf2 : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/18e1d35d22844167a15139eeb74b360b : SUCCESS in 8m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0500c7201eec4d8f9119b111891a5927 : SUCCESS in 6m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d96980b5363c4ee58eec95cd484c702d : SUCCESS in 11m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/380cbc342576427394647537bdc4020a : SUCCESS in 6m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1c43632b925c4a4a8aefb302c9f22067 : SUCCESS in 7m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2c47415aa79c4586a14c68dad1b3c36e : SUCCESS in 12m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/17031d8887cf4aad8547ecf0e3f0bff4 : SUCCESS in 33m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8df5fbe1ea334218a92f39460c0b70c3 : SUCCESS in 13m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dae41a711c0a4dff80e0e0098ced65e1 : SUCCESS in 15m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d9d8365df9254dcab18c3f5f4b6c7f65 : SUCCESS in 16m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6424cf35d794709a42f6c5d9528ec92 : SUCCESS in 7m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/686ca43804e648dd91891c651c7871d6 : SUCCESS in 19m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/232209a58ef1459cab06fdebbaec11ab : SUCCESS in 57m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3dfc286d74f84d24b2c4294a6d36451f : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":2},{"id":"789bcc718c95a78ff8825c2d46aa42c711c87501","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 23:10:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a31995368ebf63e382a66ca58d66a2e5ee4968f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 02:58:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"00c4fccdd437e4c6863bf7a428169c553fd73ec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 02:58:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea59b53663374b9b97fb851439291d70 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/342ba33d93f64f4a970bbd8cb7fe083f : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d4b9c6e38924a6c8b0f68d7835ddcb6 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a589f407a804f03ac11d956b171538d : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04aab45f55c8404589eb43f795113a84 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86111d5bf9344e8d88faefa66ee24153 : SUCCESS in 3m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/19b3f227627f49de921219047f975ecd : SUCCESS in 8m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6342a12ac46b45e3924252520a9bb1aa : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/899f133b416d4fc4beb88e3f5d2cd765 : SUCCESS in 8m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e979470481a242af8c16a700b4ecfd85 : SUCCESS in 5m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/76991fe537c340e5beeeae4c832156a4 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7dd3306357274a729084759ffc4fd78a : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bc0318ba82794ebcbd3fd5ac6410cdd2 : SUCCESS in 6m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2af33e5dcafc4607b10129c360260e5b : SUCCESS in 19m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7f3f05c6b4d94a5bb77f76741f9ad4e9 : SUCCESS in 24m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/35149bea64d840949c25de731f4518d3 : SUCCESS in 13m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa2b4c8cc3864d13b7ae51d44c53d8eb : SUCCESS in 17m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/33c3f039b61e40ca8f74eb162e53ac76 : SUCCESS in 10m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e3133cf7e264fd39e271050955e0cc2 : SUCCESS in 6m 15s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/285f3aa8addb448f8afb41114b7c69ab : SUCCESS in 11m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aea42d9fa5b34e6cbab117c97fdae82f : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4d188c6443b4a63898b4149cdd6e7ed : SUCCESS in 1h 03m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed6c7c940fae4cfe9bcbcf78d03678af : SUCCESS in 1h 19m 47s","accounts_in_message":[],"_revision_number":2},{"id":"67a1df4de5135f1ef8d33b779d3262ed7c313683","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 02:59:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4345a5e8bb22452cb0036f7c20c5e332 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2},{"id":"2c8c48ee1c2836023d449622c0e03d1d30b3f8db","author":{"_account_id":11600,"name":"Michał Dulko","email":"michal.dulko@gmail.com","username":"dulek"},"date":"2020-07-21 08:29:13.000000000","message":"Patch Set 2:\n\nFYI, in kuryr-kubernetes we had even more issues with this version as it seems to be somehow monkey-patching SSLContext and openstacksdk started to raise issues for us too related to max recursion reached.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d3ac6ba5d6b07ecf3d9cd787c5664b27c5e1efe","revisions":{"99826e064235f324875dd838fa7a8d1c9ea388a5":{"kind":"REWORK","_number":1,"created":"2020-07-20 16:30:31.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/86/741986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/741986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/741986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/741986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/741986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/741986/1"}}},"commit":{"parents":[{"commit":"237770babb13510eb5aad4813667a0265205b620","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237770babb13510eb5aad4813667a0265205b620"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-07-20 16:28:13.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-07-20 16:30:16.000000000","tz":120},"subject":"Pin dnspython to 1.16.0 as eventlet fails with 2.0.0","message":"Pin dnspython to 1.16.0 as eventlet fails with 2.0.0\n\nAs per issue https://github.com/eventlet/eventlet/issues/619 eventlet is\nnot compatible with dnspython 2.0.0 making nova novnc proxy to fail if\nTLS is configured.\n\nCloses-Bug: #1888237\nChange-Id: Ic2a38b77b5c217f769f67ed1c45d3aeb90d9cbff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99826e064235f324875dd838fa7a8d1c9ea388a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99826e064235f324875dd838fa7a8d1c9ea388a5"}]},"branch":"refs/heads/master"},"8d3ac6ba5d6b07ecf3d9cd787c5664b27c5e1efe":{"kind":"REWORK","_number":2,"created":"2020-07-20 18:25:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/86/741986/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/741986/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/741986/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/741986/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/741986/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/741986/2"}}},"commit":{"parents":[{"commit":"237770babb13510eb5aad4813667a0265205b620","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/237770babb13510eb5aad4813667a0265205b620"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-07-20 16:28:13.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-07-20 18:24:40.000000000","tz":120},"subject":"Pin dnspython to 1.16.0 as eventlet fails with 2.0.0","message":"Pin dnspython to 1.16.0 as eventlet fails with 2.0.0\n\nAs per issue https://github.com/eventlet/eventlet/issues/619 eventlet is\nnot compatible with dnspython 2.0.0 making nova novnc proxy to fail if\nTLS is configured.\n\nCloses-Bug: #1888237\nChange-Id: Ic2a38b77b5c217f769f67ed1c45d3aeb90d9cbff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d3ac6ba5d6b07ecf3d9cd787c5664b27c5e1efe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d3ac6ba5d6b07ecf3d9cd787c5664b27c5e1efe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
