)]}'
{"id":"openstack%2Frequirements~742073","triplet_id":"openstack%2Frequirements~master~I9f31247d9e233bb0fe5566db9284f3c8a6df5497","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I9f31247d9e233bb0fe5566db9284f3c8a6df5497","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-07-21 06:27:12.000000000","updated":"2020-07-28 01:49:37.000000000","submitted":"2020-07-28 01:48:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742073-1595900902260-46e369dc","meta_rev_id":"8c1dee808d6a69342c9b2cce659cfe0ed6c06c48","_number":742073,"virtual_id_number":742073,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 01:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-28 00:47:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-07-28 00:47:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 11:29:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-27 15:49:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-28 00:47:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-28 01:48:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"242da8bf8765c63acafc6794eb3bd2b10859af3c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-21 06:27:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6971ac507fc1a51a3d864429f9d2e57c941cc728","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 07:00:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3cbe2532bbfe4385bdfdfc9c5046c087 : FAILURE in 21m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1285ec02a24bff6661741422af660671d1ddd418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 08:48:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa59a30a9b14ebcb5d47b08be9d324e : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3ae37c1b2d24e45adb8a28e3ffc2377 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe46520517a14c9fb9d1031eacc3bfc0 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd981a4bb75d44f4a2e16c4ac5de5610 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7106d758aa1e431aa3e5452937182127 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77effac9827b41f18b0acd194d28b305 : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7f04e2579a1649ea927d12d9880ae2a5 : FAILURE in 5m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b30441c43110424d8760437db808e078 : FAILURE in 13m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b092d3a06c1746b59d57d5364fa660db : SUCCESS in 10m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4e89366eeb2f45ccaf93f87dbb6ca3fe : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0d9c8b43e9774b7280785cd0b957240b : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98d505815bb0467da1bfb5a936c4391f : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b0cbe1b5351349d4a80a6a99471a8225 : SUCCESS in 12m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/98b5639d281f45d6ba719774b58273fc : FAILURE in 15m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/20c97b23489d4998a2ee5e21cc48b92d : SUCCESS in 27m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/204b1241ed834fbc9c0e7e7894533af1 : FAILURE in 15m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28631201ba52424ebd2b847a5db1321e : SUCCESS in 16m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2aa4ad7962794986894ba720a045d3d9 : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/796d0bed3e214798984a5417095148e0 : SUCCESS in 7m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bec3d1e996c04424bd7412ffe1a9d7b1 : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f267f89c453420a949e23d2df09ebee : SUCCESS in 1h 02m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6faba9fa59c4c83974237d880c9c877 : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"14d421a0cc30a6f082c22b3f6b61b6d2e8507fa0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-22 06:23:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6838083d94cc594a88af9c36f5e8149f5cc86b67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 06:52:09.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/933213bd68944055b4d76e9279e9653d : FAILURE in 22m 41s","accounts_in_message":[],"_revision_number":2},{"id":"1f4e184c8da37d571613d5185669529370ce8f9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 08:49:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32aa732e7320479fb4cbed198b27e395 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e98657e88d41e38553a0bc65c9aea4 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2faaa2dce6a64943a8fefb06d0d5cf39 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2d849af6f4f4fc488153d362e6a7b88 : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9877da8bb27e49c3977aa58cdf35f650 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a245a663f9c402d85d9aa25845b598c : SUCCESS in 5m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/06fd285724e54452a8642294faa66dc9 : FAILURE in 6m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1f88f9aea7794b5da0e04ccf075a7c72 : FAILURE in 5m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/56ba2c069df346629bbdbcbf5effe751 : SUCCESS in 9m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/be884a2683a64ad597ae89cbb01d4aef : SUCCESS in 7m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1452e252853b49f59d9952b6513fc6ab : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a94b05844d54a2b8535738a1c1f8388 : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dcb386e897a64332b36034e3eee602c3 : SUCCESS in 10m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/424a0a60a40b4bb9976e7659fff7b4be : FAILURE in 13m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee764990afd84e0e99db8c9036fc9742 : SUCCESS in 27m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/059913da36f64c1bb6e358ec655f54ac : FAILURE in 17m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f431cdf422e4e069d875fae71300ab5 : SUCCESS in 17m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6dee4d21baac4476b627cced8321bda7 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5956c5085f8f4c8ea1c5ec63d14e1bae : SUCCESS in 7m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a07681894d0e4b72800d7068679969a5 : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f52a2f72282e467ca978520b0ea2b33d : SUCCESS in 46m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7cef1e799b8471eaa81e4a6d1aa9d6a : SUCCESS in 1h 23m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6395dcf1371860166123224420824f661a7d7ae0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-22 11:53:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"51f79330cc00f4d1c46ea8ce03d42e9520b00bca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 12:12:32.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/37445620dd824fe29704ac11ebe6f2c7 : FAILURE in 13m 01s","accounts_in_message":[],"_revision_number":2},{"id":"13fe219832b12aa64c3f39c3be214b41c767683a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 14:15:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b90ba67b364351a8bb6aa79c3293ea : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a7363535694a62adabc3879a852abe : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14b76583c531409a8c4bf2d7538f21eb : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82dae48b82e546718647e6018edfb661 : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f95d4f4f963144ad9ce9826df99c6e3c : SUCCESS in 5m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f57be0e8effb4b32b39b472c836e81cf : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3987a02e845e4919a7cd4405427739a8 : FAILURE in 6m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3bc038aee5214ce080591630da47ffbf : FAILURE in 5m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bb13fcb2d26f4905a3d22a00b04a3cd7 : SUCCESS in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c23b53f52ee247d1a176c28c78ea64fd : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3b7e4a933ebf466ba534f74c7be04047 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d68fca643617483187a2fcb4fb964801 : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/695b4956430a450580098f404b98818d : SUCCESS in 8m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/67062e2bf2b94618b1f66228329affd3 : FAILURE in 16m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1acab095916146d0957079bbc34aeca9 : SUCCESS in 23m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dd1e693b56144acba1ad97a45dea4dde : FAILURE in 15m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed183d279a0a4731b6508a541e052cce : SUCCESS in 18m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cbf8c3df33fd46d5af93214f35b30057 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e3ebaf28d9c4beba5f29d5875e126e2 : SUCCESS in 6m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/089bb8045bad4d529fda5750c3533ac6 : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2d8e110b7d44167b6dd3c49a0c9805a : SUCCESS in 57m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbfa975f75d941979af1988c34cc5bd0 : SUCCESS in 1h 28m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d3309c1801fadc534f6fb3d3d7667de066763845","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-23 06:42:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ca85b7e443ab023cd1c08880a896310ee4a9d6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 07:08:38.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cbee71c45fe648d2b95a3228f136a346 : FAILURE in 17m 14s","accounts_in_message":[],"_revision_number":3},{"id":"caf5c443d7eb4747ba2be2b8b5e942b3bbc232bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 08:47:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f90cc45443c40c3b8d9636606c5a1d5 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78f4e29e0d849dd920269e41c9a6473 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e85ea2b6b58d41aeaec33f8c0d4ae455 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e36df7791c649ee860e90760ad872b8 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9d30f9291564b389e806080ef42b838 : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e409eef8cac411eb55e24d32fa95b4b : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/358e2ae2706542c6b40bcef26e9461ec : FAILURE in 5m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/09b71e3fb0e14437a3c4a3a1c15ea65b : FAILURE in 6m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6a18a210064e4ac38eb9038176422c75 : SUCCESS in 10m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6655aa82e4f1409794e6a9f644a63c68 : SUCCESS in 9m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a1f6fb153e0c41ecbdd459360d854e22 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/642dc695082f41909223dd2395411137 : SUCCESS in 10m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/09c32fba824248618429da2f24019565 : SUCCESS in 7m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/442e1ce2f5684c07a54a17a8d2fd6ffb : FAILURE in 13m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3bbfe59e44f241c6a1b76da47e57371f : SUCCESS in 27m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1e8cca793ddf453ab1b3eec4648a2883 : FAILURE in 15m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d5e046fdf2f49dcab27669ffaca65ac : SUCCESS in 21m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b1ddbd690a2d45a2a13e5932da6260a6 : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83a283959a4043aaa938f62edbca5c7e : SUCCESS in 9m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/399f2c5938f84e07918eb9e0318fbfc6 : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77e49e503e2a4e748c169721fbd3b563 : SUCCESS in 56m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/268ef05642b34df48337d0953847821d : SUCCESS in 1h 23m 04s","accounts_in_message":[],"_revision_number":3},{"id":"4e716fff860c09f648c8a8f6267c78d81cce0864","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-24 06:25:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"92a57bbc7c11d9a89cf536d790e7ef77cdbe2de5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 06:55:35.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/13226a0fb54d4c3ab85ed1fb94263f4f : FAILURE in 22m 19s","accounts_in_message":[],"_revision_number":4},{"id":"e256306283065eee27017f5c2b60abd2a822c068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 08:02:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9eefb0eba04c22b1f6d7e23db88faf : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/643ff61e040943e3bdc2670d501fbfe6 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f939369932fb423a9b745b83035ad3f7 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a07f173befed4c24bbb661c4c1aad8fa : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13094df8d00c428a8bdc90b777aeb751 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7477246ef3804ded9a7d8c7babc6c592 : SUCCESS in 5m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/070a45ba176f40c2b032f3cae1184702 : FAILURE in 7m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a2cdbedbf17403fae5e075a3959dd71 : FAILURE in 7m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aae5e9842465413cafc1abfa97194432 : SUCCESS in 10m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5adc3bbc365e4bc0ae0acf27617f9d6b : SUCCESS in 7m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/266c1e9363af4a7dab73699263971b5d : SUCCESS in 12m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c8751d9ef4847bcb17c3f6fcc709cb9 : SUCCESS in 9m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/20299f8e8bfc4dacbc031ba08f1b6696 : SUCCESS in 8m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6a0e9319e24d4018983fa403d8a59d9a : FAILURE in 15m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4abe91ffe1234ffca4f12a041021ef71 : SUCCESS in 19m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c838eb1f1620412d9a4b04aece312ca9 : FAILURE in 18m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6929e39300140dbb9a859b246f8e49b : SUCCESS in 20m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aba344356f1840e290c3abe53f44c333 : SUCCESS in 8m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f0f6f7b653745f5b2aafb249acb568d : SUCCESS in 9m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e5a57cf99654760a621fe39db30725d : SUCCESS in 14m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed90850fef5b45f1b97562dfe2cc5c62 : SUCCESS in 49m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f79ea5ce27ca438e944a48300a490b14 : FAILURE in 1h 07m 06s","accounts_in_message":[],"_revision_number":4},{"id":"30b532da9129c454329f74eebcbc5de1d95e1c77","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 11:29:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1542abd2daee5ee6135370df44bb02f40ecdc5cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:47:09.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a1b5c32b303248079522c92b7b7aa2ca : FAILURE in 12m 40s","accounts_in_message":[],"_revision_number":4},{"id":"2ce9c83b042bc12ae07c6ed4ad33a52eeb98bccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 12:46:22.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd58a14381641b1b9aa92b2fcf825b9 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b6d8e763a34616b56244e9bb734de9 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c99a40b9900495a84eda80f8ac8ae12 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa77c37e0a1148e4926c069dbb6a946e : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/753b7dcda1fd4909b7289d741d08dad7 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3ce9d138d7c744ab92bdbe99563592d7 : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1086bd658c15404db1e8735925e8a2c9 : FAILURE in 7m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/22463b012e024179af2a996720aa261b : FAILURE in 7m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0ed07ea4b4ab4b43bd581beed566bddb : SUCCESS in 9m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/599fa99578d84266bb8c5f2232855fe1 : SUCCESS in 4m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad013946d62e4d2d9036fae653b933e4 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e7e81f1519054b3a971947939e125a11 : SUCCESS in 6m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d1d27018e8544d2ba5d24cfcf12833bf : SUCCESS in 7m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b9ad11fb74644311841c211190f946d1 : FAILURE in 18m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/333473c076f04b73b9220094cf983060 : SUCCESS in 34m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/38ca0e0a139c4b4d9e6b4b8513a8a039 : FAILURE in 11m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/266927eeb6e74b5bbb4b1c5da2cc7886 : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6e63949a74f344e5a04daec56acc7637 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f05b56e68d054c2083562ea349e32cd2 : SUCCESS in 6m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74e8cb4d3f834b76853db450e2cd2af5 : SUCCESS in 10m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52b5ad99a2394f7c8f5c00587f511174 : SUCCESS in 41m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b89e90e520694c0dac6ae3363ec80bf2 : SUCCESS in 1h 15m 56s","accounts_in_message":[],"_revision_number":4},{"id":"5b48b223acba5816e36960cee4c0aaeab1a310a0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-25 06:25:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d004fca578ab186f2c41ee66e196ceb54a951e2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 06:51:09.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8f75a235ec0c45fc9762c6af73fbfcec : FAILURE in 19m 20s","accounts_in_message":[],"_revision_number":5},{"id":"0dd2bc5f95cabaf417fb13fbee70a8649aec3fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 08:10:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caefd8199da94c74ad6eeab218a01f60 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3c239128244649b83c9d8c82e208cd : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6c724ab92c84b4b83b19bdef5ae425b : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3143f7bdc0b4dddab7866c8036ff5d9 : SUCCESS in 7m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb724513fd794a84af153af62919cefb : SUCCESS in 6m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/484633b56e3f4bee911a84051de75a85 : SUCCESS in 5m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2043a5bd11e146d2976325df2c94c3f7 : FAILURE in 8m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/63a71dc4937d4c088f71ad628d3b6c6d : FAILURE in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3ff7dbf86bcd43d593d38292cf0bb43f : SUCCESS in 11m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/635d9436cab54b2590fd0b2ec31154a7 : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/89fa3760a80b454ab44214504637fc5b : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8525019f8580498081559502b29acd1c : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f48f8bef05f94884b5f37a876a2f4f9e : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8465febcb77846a0804d2310beb5bf7b : FAILURE in 14m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6520891c82404278b6f4c45f4d3694f9 : SUCCESS in 23m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a4f90ffa269e41b985a1d39d8d23fba6 : FAILURE in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/504d8207faee4fb5844b263a1bf6507f : SUCCESS in 20m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f7c017e9f5a54875b7b5c3e608bcad38 : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/292e41cd69454b7291402ce423e653c5 : SUCCESS in 9m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/507cf5a0bb494bcd86e97a79e713f8e8 : SUCCESS in 11m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9b63ea159924e358e2c2e0bca353985 : SUCCESS in 49m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0dfbc4aab9f45158d322efa788a287c : SUCCESS in 1h 12m 29s","accounts_in_message":[],"_revision_number":5},{"id":"3ec7a17bbfefcb030c39c1aa68e25fc697c8b840","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-27 06:31:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a2a4d83fb3795a5405d7640cdb5a424e4cdc1954","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 06:57:37.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f8eb59bc4acb4ca18154c90d0658238f : FAILURE in 17m 45s","accounts_in_message":[],"_revision_number":6},{"id":"b4c18a808a1505ecbedb14816ce1dfd273245700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 08:45:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f189e403aa4236a0dc6a3b2500856e : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c47c9fcc0f44441b8511892357472dfd : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19a4f5052aea430fab9e6356bea0ed59 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f169af2d2d6c4d1f8393b8ad70ad07e2 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57e252f5591b4f5796a57f25b55cffff : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e140209b9dc49039da072c52549c707 : SUCCESS in 4m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d1cd25e3ee44bd2bba086ecfa1eddb0 : FAILURE in 6m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d55d60a0f28f48ad8f35a5b99f0c6e0d : FAILURE in 6m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0cefce9e0d724bfcbec44fbc84393207 : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3c8742dce228408080fe442312efde5f : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/adf7438381dd4011817386df2a387a58 : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b37dac7721d741a695b61966337a9131 : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/649641368b334c788581cbb05c645c9b : SUCCESS in 8m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e5cc663c218748faae38332e565696fc : FAILURE in 14m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0734c3dd5c9d4fb38b5f7c3b8f4ac953 : SUCCESS in 27m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1f43b31f79024d9ba089686b8808ff67 : FAILURE in 14m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/556979963e7941a5a76b0cb320c4fd08 : SUCCESS in 17m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2ea52572a6284d10a938a887a0491671 : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1534d62e7f0d462e8d6e97e767de56eb : SUCCESS in 8m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e448f5350e5f45d294de32a16610bdfd : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ffb06f50cac4596a311d9856a7c5a90 : SUCCESS in 51m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b29affda123e415fa958b671a0751df9 : SUCCESS in 1h 22m 23s","accounts_in_message":[],"_revision_number":6},{"id":"deb72f1f98098bb99aff45587580589787f5500c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-27 15:33:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"518ae89175303a21a5ecb46c7b0575fd64632127","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-27 15:34:41.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4e4e3e9cac8916ffdfd192da19b3246bd257868b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 16:24:43.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"7f31023017026c07fbb608b8f70a6075db119f2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 18:39:25.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b30ac997c024ebc93c10814ae3ffba8 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad978c7db424430bc3b81a47ed5b536 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d4bed4735e440c1a897f542a01afbbb : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3d250fcd500455abecdda64677f366b : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14f8b4f5812d4e9eac31ab0bcb07988d : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff948ec5a03240c6977244e6ea1c9299 : SUCCESS in 4m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7a24448720ee4630b8330fa8de4a0995 : FAILURE in 5m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ab1a5cc3fcd548c8b669e7db4f521963 : FAILURE in 6m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/53bc83ed39d047f7a78ea63d830db552 : SUCCESS in 9m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4004b7b013bd47beb5a60e52beedde31 : SUCCESS in 6m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e869feea086e46ee95b5d90ede8e0f06 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f374ac89e0434cb7bf5d1b07b57cfea2 : SUCCESS in 9m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0c0d8917b1b44384bdda2e3e755e14a2 : SUCCESS in 9m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f6a093d60544affb1c2fd7b37718fdc : FAILURE in 14m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/be903f8dd7bc4be2a35c6835449ea5dd : SUCCESS in 25m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fe67c4cb4bd548c6876c9772af6b7feb : SUCCESS in 15m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20b9d6e71be44784a996734e22f17b2a : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/38dd771bbfda4eac878c0e5b636429f7 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/467c3752dbc348cb8321a2735f0f8b62 : SUCCESS in 6m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f8881c88998043e2af65e900f8af6210 : SUCCESS in 9m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0ce86892d4e429c8edba26da85e10a0 : SUCCESS in 55m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d67b56a72d70422b82bfe00bdad3b538 : SUCCESS in 1h 17m 14s","accounts_in_message":[],"_revision_number":8},{"id":"971051f6ea34a283d5e14eec54af5c6b7dd54731","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-27 18:45:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2752839e6fa7bb8f6f4b6a08b62a46f8402eca7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 19:20:15.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"9b9f8200e58f028f53b5c31b8c976e7f052f64db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:52:06.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86fe7f95a91e45bfba5d9dfdb2357719 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef61820eb5245e8ab4d5b0c771e6ec0 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38a1b8d7944e45e88941bd273abe8c6b : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/187a32fa1bac4dafb54e0f6ef9975c8e : SUCCESS in 8m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3798f521b1a04368be7882a2fd7ba3b2 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/339911814d6f4bfcac86decf9d429f99 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f83d903e986240ba896d767732847880 : SUCCESS in 7m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a38b9c8c7078401e8cadb050cacd34e3 : SUCCESS in 14m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b9f74ce7742e4544b313178bc0604daf : SUCCESS in 10m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3c6b27ec8464400289550845726a6e2f : SUCCESS in 5m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/96c54f29ea24415cbbd0104b82ae1592 : SUCCESS in 13m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd0fc7ac4c5b4a78bd0c16567c886590 : SUCCESS in 9m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c477c80941734a2083cb597533e8ef9d : SUCCESS in 8m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/93d5bb30152345a99a4e7dc572029493 : FAILURE in 15m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f25d3aaaf7e3430698b0eaaa6842d1b1 : SUCCESS in 26m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8427d228de654a2a8798e62892955e17 : SUCCESS in 16m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26288ab9f02f44958f474723b6b698c3 : SUCCESS in 15m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/af8c792fd9a840a28845c559ab33d7ce : SUCCESS in 9m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fbff84fb732c411fb7077900846d969f : SUCCESS in 8m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/be84d2cafbb349baa539fd2e7f19df38 : SUCCESS in 14m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f3a22abace6488381227beb139d818f : SUCCESS in 59m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ec88eaa62774e5682987dfe4aacd13c : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":9},{"id":"e3effaaecad00555b94bcf6c30482f4d9dd18d91","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-27 23:07:00.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5a15756949289934cfe4864d211eb62b4fbc9fa5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 23:13:22.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"6db39008a7d4f8c2279fd0e98e43836d4607d404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 00:40:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5151a20dff6c4681bcdfaeee6e466ccf : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efdce87d07b944f5b9ecea4eec290d08 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04e0b202d45e4c668ef49b2116964264 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9953d0a4aa874b4199641a41745b5fc5 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/497638dcada848659ad574a01a8ac4b7 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1933de9ba2774ab48f990c084173273e : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fca59ad1ec76442e98c71538c53ad8a4 : SUCCESS in 17m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ad2fd087482436cbad2e45fab5a67b0 : SUCCESS in 10m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ca74b465b3414b29a88d3d10fea6eb1d : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1230640c0cc44230a064c47a879992eb : SUCCESS in 7m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e74ea59dcc98457199f200789f01baf0 : SUCCESS in 13m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/598c1d5ac35742bda1745f39c8992e4b : SUCCESS in 10m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a2b216fb87b142f8950446bba05ef32b : SUCCESS in 8m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/66d08849afc44d989e5fa06bcb39ab8c : SUCCESS in 34m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa591b200e0249889befa959a93823f9 : SUCCESS in 27m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2bb34f8696ec4065ad043c3c1eb3ee31 : SUCCESS in 16m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/39610b3e9c2342e787ce8d4b00edfc9c : SUCCESS in 19m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1845718a37a34149a8868a28efce2977 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/83a513b34e2f46a4bfc072ca430b6c41 : SUCCESS in 6m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a2ee6512e3949e1a80a3c3977aae936 : SUCCESS in 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb5c6ae519eb4cd9840a9deca380a941 : SUCCESS in 52m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c39c7230d0864b7bb8da59b663d38e28 : SUCCESS in 1h 22m 54s","accounts_in_message":[],"_revision_number":10},{"id":"4baaa6ad3ad220a4e70409ff92de31990b9967a5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-28 00:47:08.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":10},{"id":"5c4e9fcad71df3c456329b30af63f4c13520162f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-28 00:47:14.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"5ba487f624c34a5e46531be3270b16495b0b4b81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 00:47:29.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"d3b712cf3214f452efa9ce1bef5bdde632d2fbb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 01:48:21.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115c7ebf7d194defafa0a996eb0e08d7 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b183026da24e5298796731564171f8 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81af4737ea5341048edbd3695798c164 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9db80b15a89440da8f1514182f4924d3 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa25d7ddfcda435ab16483e2fca6a6f0 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e48062330fc24ad78c8c21d57edd7b7f : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0d47d50ed47745afa684a107cb66f803 : SUCCESS in 9m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e2ae6ab069ec4aebb1c4397d94d451b3 : SUCCESS in 10m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1d7e9dd083eb41b0bd1db7e6c2570825 : SUCCESS in 8m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6fc4541f71ca41468f96320e01009b05 : SUCCESS in 4m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad3a98cc4e9d4c60866c834c2d7b9d1c : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b1c865215cbe44e1ab6df37d0c95c5a3 : SUCCESS in 7m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1cce5f57a9b440cdb26c6f30631a71cb : SUCCESS in 6m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e57a05f1855e41d6b0aca35f46870983 : SUCCESS in 13m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5723b12988c8432482b59acea8ef73c8 : SUCCESS in 22m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab734a664fae49c69cf11e1ab01b2dec : SUCCESS in 12m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7aacbef970aa450ab0fb2701dccc48d7 : SUCCESS in 19m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/99d1209ee6d6443eabd27e9f07f02d51 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf6e0b7e6ff44d8cb88596ae744c3b29 : SUCCESS in 9m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a147b656e26240f79f966f805ebf935a : SUCCESS in 8m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6848daa7ef414a94bdb6524202407cd7 : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a43355a2fd141259f02471710539342 : SUCCESS in 52m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4b99d49c4284d5b87bbe66bcfc29c02 : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":10},{"id":"b3bda004e33eb9ca93f5ab30a600064521c0c402","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 01:48:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"8c1dee808d6a69342c9b2cce659cfe0ed6c06c48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 01:49:37.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a548bc44dd4310ac529bcdd3b972f1 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"7654c5464598630962767af587ab11739d771e09","revisions":{"e22c5907fa4b3add227f36ca55f2ea169029b30c":{"kind":"REWORK","_number":1,"created":"2020-07-21 06:27:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/742073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/1"}}},"commit":{"parents":[{"commit":"30109be02aec75349c502e98417deea9c2806846","subject":"Merge \"Pin dnspython to 1.16.0 as eventlet fails with 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30109be02aec75349c502e98417deea9c2806846"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-21 06:27:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-21 06:27:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e22c5907fa4b3add227f36ca55f2ea169029b30c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e22c5907fa4b3add227f36ca55f2ea169029b30c"}]},"branch":"refs/heads/master"},"465c4e4ad896b6639ba7e6af8d21a617527e290c":{"kind":"REWORK","_number":2,"created":"2020-07-22 06:23:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/742073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/2"}}},"commit":{"parents":[{"commit":"33b69dbe9e6db48ee844c9fd9988df561f702641","subject":"update constraint for tooz to new release 2.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33b69dbe9e6db48ee844c9fd9988df561f702641"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-22 06:23:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-22 06:23:02.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/465c4e4ad896b6639ba7e6af8d21a617527e290c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/465c4e4ad896b6639ba7e6af8d21a617527e290c"}]},"branch":"refs/heads/master"},"05dfc7306962dce912622f18b4fafcd129726652":{"kind":"REWORK","_number":3,"created":"2020-07-23 06:42:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/742073/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/3"}}},"commit":{"parents":[{"commit":"1b0509b52875c480d653870fbea842afe488ae8e","subject":"Merge \"update constraint for validations-libs to new release 1.0.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b0509b52875c480d653870fbea842afe488ae8e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-23 06:42:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-23 06:42:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05dfc7306962dce912622f18b4fafcd129726652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05dfc7306962dce912622f18b4fafcd129726652"}]},"branch":"refs/heads/master"},"88bae228c268ca4edc9534095d0b261874f02c25":{"kind":"REWORK","_number":4,"created":"2020-07-24 06:25:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/742073/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/4"}}},"commit":{"parents":[{"commit":"614d950132849f2940b618d776fad77bb936c3c8","subject":"update constraint for oslo.messaging to new release 12.2.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/614d950132849f2940b618d776fad77bb936c3c8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-24 06:25:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-24 06:25:28.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88bae228c268ca4edc9534095d0b261874f02c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88bae228c268ca4edc9534095d0b261874f02c25"}]},"branch":"refs/heads/master"},"1f38c608a5c8103c7acec9ac9cd905d829de692b":{"kind":"REWORK","_number":5,"created":"2020-07-25 06:25:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/742073/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/5"}}},"commit":{"parents":[{"commit":"de8bb104fb0cfd9503e70ad2a986cf0d54833031","subject":"update constraint for ironic-lib to new release 4.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de8bb104fb0cfd9503e70ad2a986cf0d54833031"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-25 06:25:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-25 06:25:57.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f38c608a5c8103c7acec9ac9cd905d829de692b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f38c608a5c8103c7acec9ac9cd905d829de692b"}]},"branch":"refs/heads/master"},"2fc6732cbb5cc509c3cb92295eb226220a524308":{"kind":"REWORK","_number":6,"created":"2020-07-27 06:31:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/742073/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/6"}}},"commit":{"parents":[{"commit":"de8bb104fb0cfd9503e70ad2a986cf0d54833031","subject":"update constraint for ironic-lib to new release 4.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de8bb104fb0cfd9503e70ad2a986cf0d54833031"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 06:31:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 06:31:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fc6732cbb5cc509c3cb92295eb226220a524308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2fc6732cbb5cc509c3cb92295eb226220a524308"}]},"branch":"refs/heads/master"},"0369046d9d4d10456ece82bb52319236b110b12f":{"kind":"REWORK","_number":7,"created":"2020-07-27 15:33:21.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/742073/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/7"}}},"commit":{"parents":[{"commit":"de8bb104fb0cfd9503e70ad2a986cf0d54833031","subject":"update constraint for ironic-lib to new release 4.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de8bb104fb0cfd9503e70ad2a986cf0d54833031"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 06:31:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-27 15:33:05.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0369046d9d4d10456ece82bb52319236b110b12f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0369046d9d4d10456ece82bb52319236b110b12f"}]},"branch":"refs/heads/master"},"4d7269756170b35751782fd3daf9e42ae5782b9a":{"kind":"REWORK","_number":8,"created":"2020-07-27 15:34:41.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/742073/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/8"}}},"commit":{"parents":[{"commit":"de8bb104fb0cfd9503e70ad2a986cf0d54833031","subject":"update constraint for ironic-lib to new release 4.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de8bb104fb0cfd9503e70ad2a986cf0d54833031"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 06:31:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-27 15:34:18.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d7269756170b35751782fd3daf9e42ae5782b9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d7269756170b35751782fd3daf9e42ae5782b9a"}]},"branch":"refs/heads/master"},"dc7a1855ed573fb9c6b4a81a868ce55a1799a5ae":{"kind":"REWORK","_number":9,"created":"2020-07-27 18:45:01.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/742073/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/9"}}},"commit":{"parents":[{"commit":"de8bb104fb0cfd9503e70ad2a986cf0d54833031","subject":"update constraint for ironic-lib to new release 4.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de8bb104fb0cfd9503e70ad2a986cf0d54833031"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 06:31:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-27 18:44:21.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7a1855ed573fb9c6b4a81a868ce55a1799a5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7a1855ed573fb9c6b4a81a868ce55a1799a5ae"}]},"branch":"refs/heads/master"},"7654c5464598630962767af587ab11739d771e09":{"kind":"REWORK","_number":10,"created":"2020-07-27 23:07:00.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/742073/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/742073/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/742073/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/742073/10"}}},"commit":{"parents":[{"commit":"de8bb104fb0cfd9503e70ad2a986cf0d54833031","subject":"update constraint for ironic-lib to new release 4.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de8bb104fb0cfd9503e70ad2a986cf0d54833031"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 06:31:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-07-27 23:06:31.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\npymysql held back for keystone issues\n\nChange-Id: I9f31247d9e233bb0fe5566db9284f3c8a6df5497\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7654c5464598630962767af587ab11739d771e09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7654c5464598630962767af587ab11739d771e09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
