)]}'
{"id":"openstack%2Frequirements~742448","triplet_id":"openstack%2Frequirements~master~I1f0b07423a4ef80ca852d6244d0cd87dc2534f3c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1f0b07423a4ef80ca852d6244d0cd87dc2534f3c","subject":"update constraint for validations-libs to new release 1.0.2","status":"MERGED","created":"2020-07-22 14:28:52.000000000","updated":"2020-07-22 21:01:26.000000000","submitted":"2020-07-22 20:59:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742448-1595451572468-1ee57552","meta_rev_id":"ea25d3c6560c9a79bf107533163fa600f965b9c0","_number":742448,"virtual_id_number":742448,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-22 20:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-22 14:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-22 19:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-22 14:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-22 19:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 14:34:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-22 14:53:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-22 19:25:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-22 20:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0f67b9b302ce0e1986337fdf83212debd1e34b2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-22 14:28:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7afa9ecd5ba4ef982e7fc0bde7141b7b5c5ebafa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-22 14:34:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e5c41773a64ebaeb19f6b5bbcb19a2755dfd28d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-22 14:53:03.000000000","message":"Patch Set 1: Code-Review+2\n\nWaiting for CI","accounts_in_message":[],"_revision_number":1},{"id":"28d3d16c6f07fadf55ff1596920d5dc73973a7fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 15:10:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/523b9a162a994551b3d90f2b9e3a3dda : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0dc31ddaea3f31ab3c792812d8906feda9a7e257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 18:20:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af05cb2e7fe741ffbfea65f9e13ae1ef : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ee3022602542fab792bb3a49badc22 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2148cd24efcd4433bbb7c203fcb562eb : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aa63b300aed4f458f372b6b5d5c03b5 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b83c81be8818464db4229fd7dfa7385b : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/683b74e01f474d50bcfa1a9004f2d85a : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6a2c0d54581f443cbbe2e349a5c4e809 : SUCCESS in 8m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0adbdef64963418a944de6d038e87700 : SUCCESS in 12m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d253de9e164741c2a708da2ec943d121 : SUCCESS in 10m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/108a514e36a646eca25cc2f16cf66309 : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bc1a62d2ef124ea2a1a581aab4ed7b20 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d69db10975974982acf25ebe6758a24c : SUCCESS in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ea2e6c41ef254dc49e10d44a0cdbd3c6 : SUCCESS in 9m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cab1b97ee313461494be97eb1bb102db : SUCCESS in 14m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6e2d18e002084d5fb0104cad53304a9e : SUCCESS in 30m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/105b57be51f04fb1b4d7f73c617548fc : SUCCESS in 13m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1d805925d2b440679ea49e650a23aff2 : SUCCESS in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c19902030056455ab48516cc82e361eb : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/12f6c15c127847ffa76387651c15a008 : SUCCESS in 8m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a09af089617448308c7039d4a3c7d21a : SUCCESS in 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49808fd2d6424dabb70c4f0fc7f60ba5 : SUCCESS in 1h 06m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/066322edbc594ad78a14f778b3865579 : SUCCESS in 1h 34m 17s","accounts_in_message":[],"_revision_number":1},{"id":"adc8beabbe39c1f8ed3930fba531a5c826b85910","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-22 19:25:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4b32d83405a9a3414e672e6c03185c6a28daab8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:25:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"414931425e0462f1f88c3952713bcac5ae811a58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:59:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600d894e8d9b4b74bc1fd6c84f3f86f5 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201da7af46db432ab1aed97263652360 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/960c72d5349d41b687486122eb567919 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fb9121bd34847128673f17c29dfe7b8 : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51da9559f34248db9310b56c2c718d65 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6731e4dfbf64104988e28e61665b552 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/22bce602d0da4291be7de2a28961f11a : SUCCESS in 10m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5fb1098564c14780a1c415b10cc0fddf : SUCCESS in 15m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e9da829119cf4ea99a681f5e4f3ee0f7 : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e784249d69684521bdaba2e86767f0f2 : SUCCESS in 6m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/787a0416db8f4848acfbc2b4a2d86d77 : SUCCESS in 15m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b320b15bb0554c4aafd51db035b0bd44 : SUCCESS in 7m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/537612bed4224281bd1b6353d7468e00 : SUCCESS in 8m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3bab800c9996408f8d83085684d429e2 : SUCCESS in 18m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e350a2b0819143c6be132999b800919d : SUCCESS in 30m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/daedb6b5af7149119e3a665eaebffe10 : SUCCESS in 15m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b5ad20783e14a71930ead622eccffb7 : SUCCESS in 18m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/38726b2f547e4249880770c617decbec : SUCCESS in 7m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86ade8b46a3b493da9ae57823f20762f : SUCCESS in 7m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3da06ea1fb1346148d350f435e7fd739 : SUCCESS in 10m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/60620a3184b245c083307dddf27e9336 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3219eeaf5720434fad250b6a6730a467 : SUCCESS in 52m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d62980392024df197a5e0bba1a490eb : SUCCESS in 1h 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"97a3bd28a545d0dca699b2fb03dee48c398f49d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:59:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bb6a7759fc30d67fe7ca621cea360f17bba6ade4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 21:01:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e947ff90a6fb407eba6eef101c7adfd2 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d4f7d377e4305a967e84c88d07d0e92f4259cd7","revisions":{"4d4f7d377e4305a967e84c88d07d0e92f4259cd7":{"kind":"REWORK","_number":1,"created":"2020-07-22 14:28:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/742448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/742448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/742448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/742448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/742448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/742448/1"}}},"commit":{"parents":[{"commit":"33b69dbe9e6db48ee844c9fd9988df561f702641","subject":"update constraint for tooz to new release 2.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33b69dbe9e6db48ee844c9fd9988df561f702641"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-22 14:28:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-22 14:28:49.000000000","tz":0},"subject":"update constraint for validations-libs to new release 1.0.2","message":"update constraint for validations-libs to new release 1.0.2\n\nChange-Id: I1f0b07423a4ef80ca852d6244d0cd87dc2534f3c\nmeta:version: 1.0.2\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Mathieu Bultel \u003cmat.bultel@gmail.com\u003e\nmeta:release:Commit: mathieu bultel \u003cmat.bultel@gmail.com\u003e\nmeta:release:Change-Id: I18ab5d7afa350746ba2310a219755f07ef957b06\nmeta:release:Code-Review+1: Cédric Jeanneret (Tengu) \u003ccjeanner@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d4f7d377e4305a967e84c88d07d0e92f4259cd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d4f7d377e4305a967e84c88d07d0e92f4259cd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
