)]}'
{"id":"openstack%2Frequirements~742530","triplet_id":"openstack%2Frequirements~master~Icc3b05558f8caec395cd8b194cd99b4db2e8636a","project":"openstack/requirements","branch":"master","topic":"osa/os_congress","hashtags":[],"change_id":"Icc3b05558f8caec395cd8b194cd99b4db2e8636a","subject":"Deprecate os_congress project","status":"MERGED","created":"2020-07-22 19:58:03.000000000","updated":"2020-07-29 11:58:42.000000000","submitted":"2020-07-29 11:57:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742530-1596023824100-ae122b9c","meta_rev_id":"e9b1c34de59dc825b79b43347def4fc38d145feb","_number":742530,"virtual_id_number":742530,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-07-28 11:58:05.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-29 11:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-22 23:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-29 10:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-07-29 10:14:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 23:31:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-29 10:14:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-29 11:57:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e60643df9bace6f505ea2f1481bb73dccb2c195f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-22 19:58:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"563870eeacf0da38d59d3ca716ffafc3a8686e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 22:20:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43550239710d4a58832e81c3c79f97b2 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/070d6988f3a54575b92a8ce7a1c3a49b : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c036301006d9403a9dfb6337fa503d40 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/953e8bf55c75458eb21ee274ba64c18b : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/381547aa34c94ffcab48e6cf67fe6715 : SUCCESS in 3m 57s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/68f7c0771d704802a62aac82167bb8ac : FAILURE in 4m 03s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c496f1e33994b5790f25bf0afcfebf8 : SUCCESS in 55m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd1a7bbd1200459ead1e8f4df9188d4b : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ab19b3f47b43a9e8f3a2225f85e1f5bb1e48815a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-22 23:31:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4328fc7ff1495da4e8410a37efdc54e9fe38e578","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 11:33:16.000000000","message":"Patch Set 1: Code-Review+1\n\nI think we can go ahead there but I\u0027m not the more informed person on this topic, I seen lot of discuss and patch related to congress during the previous weeks so this is why I think it\u0027s ok here but I prefer to only +1 to allow other cores to take a look too, especially Sean who if I remember correctly submitted few patches related to congress.","accounts_in_message":[],"_revision_number":1},{"id":"e4a0e2eee6427b2ca9a9eaf4c2dbd55d88dc198d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 11:53:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a791f94c5dd4a6d69789c04726754faafa4d7696","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-28 11:58:05.000000000","message":"Patch Set 1:\n\nI think setting such dependency will cause circular one, as we need zuul to have noop job for the project for merging https://review.opendev.org/#/c/742520/","accounts_in_message":[],"_revision_number":1},{"id":"d35da749b4a252a8088e73c7bc3b8173789debbc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 10:14:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n\u003e I think setting such dependency will cause circular one, as we need\n \u003e zuul to have noop job for the project for merging https://review.opendev.org/#/c/742520/\n\nOk then let\u0027s go with that","accounts_in_message":[],"_revision_number":1},{"id":"192e08fdd43e16c981173239ade77e518d5a3f0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 10:15:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db4bb285f83b6ec36282a8685eb34ea9687e6a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 11:57:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8d4ff7c7c04758b99c0e89cfc13349 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23eb95ec95014928babda2e0b6d4751e : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46e50283e25544fabf108c5c5fd2a130 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbff46ca4664485f94f445cf7c96ba9e : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e49ad130391b4bf39f7c35373fa8025b : SUCCESS in 3m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89ee3219a74f4e8384d89460106c70ed : SUCCESS in 52m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba9379314223470ca8f7f052ab0d9cd5 : SUCCESS in 1h 32m 27s","accounts_in_message":[],"_revision_number":1},{"id":"007b6db48a7f7ebd6f557309ea33d0a4afd597ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 11:57:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9293ad662666d857695e2897d26a85450b853d18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 11:58:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/325b8610721f4b798cbc2d1154765ac4 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41837ddc482e67c10c3e6651c3fb5d1661e69764","revisions":{"41837ddc482e67c10c3e6651c3fb5d1661e69764":{"kind":"REWORK","_number":1,"created":"2020-07-22 19:58:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/742530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/742530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/742530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/742530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/742530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/742530/1"}}},"commit":{"parents":[{"commit":"33b69dbe9e6db48ee844c9fd9988df561f702641","subject":"update constraint for tooz to new release 2.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33b69dbe9e6db48ee844c9fd9988df561f702641"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-22 19:57:52.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-22 19:57:52.000000000","tz":180},"subject":"Deprecate os_congress project","message":"Deprecate os_congress project\n\nChange-Id: Icc3b05558f8caec395cd8b194cd99b4db2e8636a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41837ddc482e67c10c3e6651c3fb5d1661e69764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41837ddc482e67c10c3e6651c3fb5d1661e69764"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
