)]}'
{"id":"openstack%2Frequirements~742615","triplet_id":"openstack%2Frequirements~master~Ibfdae9e6e2d85f2964733ff7e94ab2f42da6ebaa","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibfdae9e6e2d85f2964733ff7e94ab2f42da6ebaa","subject":"update constraint for osprofiler to new release 3.3.0","status":"MERGED","created":"2020-07-23 09:15:06.000000000","updated":"2020-07-23 17:35:42.000000000","submitted":"2020-07-23 17:34:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742615-1595525646924-a26e969d","meta_rev_id":"f2f90671eab66b042557e8c714855fc96cf021df","_number":742615,"virtual_id_number":742615,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-23 17:34:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-23 15:24:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-07-23 15:24:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 15:24:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-23 17:34:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45df698cea190cc1b0b15f35db222de118fbb70c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-23 09:15:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"947423812311f11457ec55b029ad38b16d5b03bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 10:09:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f6912cbcd42a44eab951320467fec43f : SUCCESS in 48m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d677507c46e9eb9615e043495a92eb3e0a7352ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 10:42:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b37bf7b15b24c0bb40f36b2ce749d55 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fbbb61a46b4404387a1587c8961ebb6 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41588c42128040f2ade8c1a72ffd8aa5 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eceab00d5b3d4b6098a2ff9889f16e8c : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b89bf5872b54549ae091b6c440cb1e0 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/849c2eaf27ce4cf8a4c486aa93e5ddd2 : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b2e9612c8a4d4ac2a0aec9bd296ac779 : SUCCESS in 9m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/17fcb800f55a45c9bd6e647b181ab1d1 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/663ccb20dc864be0877c82a2b230a7a5 : SUCCESS in 8m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/de1953e932794436aa45e17660e3a64c : SUCCESS in 5m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1b8ec30fc75c423782fd868b7dd710fd : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b30a4b63a05445ba892d34f126a0a5a8 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0e836251964b403984f26a3b3bacc480 : SUCCESS in 9m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/60305207ea5449e9908c6eabe939a4a3 : SUCCESS in 18m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0d6141cb3c724bf286a680fa50dc617e : SUCCESS in 24m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/da672f22d5a248e0857bf2d3b284eac3 : SUCCESS in 16m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5dbfbc8e883f4e09a618e0258eb741d6 : SUCCESS in 17m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bc419aeecfb84f28b02b021b983a014a : SUCCESS in 8m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3cdae27e417f409189fbe0df6ef2f918 : SUCCESS in 8m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/22eb304a02b64b96bb34b67c5eee61a9 : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e775c5b820b94aed9506be87daa04755 : SUCCESS in 53m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/de2c16bbd9fb474294ff53d0263a7bf7 : SUCCESS in 1h 16m 26s","accounts_in_message":[],"_revision_number":1},{"id":"01de10162a8008693b1557be47ddd0db34453a1e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-23 15:24:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f991ca55362e9c7a995892b8754d0ac448aaa564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 15:25:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fa64a6c6f7f3b8ef8039b5a01f820d33736b5ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:34:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e3ccbf452aedb6e24e7d124df6e2de0362923024","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:34:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e8fe3fee9a440f8395d76e2fefbf0e : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb732ca9d4c43149d3f4af46f66fb67 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/190f704d34ca41278269c615a588bc0d : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a20df443a864e7189781206fa4f55b1 : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d58255af557f453c8d0b459d254763e3 : SUCCESS in 5m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d241fbebd92e4fcaa15a405cb765eab2 : SUCCESS in 6m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ee39794462048f5aeec19e51a8f60e1 : SUCCESS in 7m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94d46e55cf844e8e8e24f6e9c206c36b : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/90dc29a6475b4ebeaf0c526898f72059 : SUCCESS in 10m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7efd39773ef743caabafc7a1a73029db : SUCCESS in 7m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/558fe82a4cee400a88f604c37764de7a : SUCCESS in 13m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc733179f9e14aa49240399fc97a9fd1 : SUCCESS in 7m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/24e86ff4bf0c4e5090ba7795959450d6 : SUCCESS in 8m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b11d27fe1a8a4ab894bc413ca7f835a6 : SUCCESS in 17m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b8fd3e82508d4fbca4469d72d302aa20 : SUCCESS in 27m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a4667284ba00442390be7d8697c6b585 : SUCCESS in 11m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2c99b5d19d6412183db8cf703f3d789 : SUCCESS in 18m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b99cd27548994639b71735dfb0fa0943 : SUCCESS in 7m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4cde1f0483774c2fa280302e538e9e13 : SUCCESS in 8m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f4678c87471445378f9b652f62b72f16 : SUCCESS in 11m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/be623f21619640e49f4f2770b21d6b88 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a21e032228ba4e5b94c9de58d4354249 : SUCCESS in 44m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b14d4e390564429585a965e08395159f : SUCCESS in 1h 28m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d6e1ec769a20d7e57765e2f81eda9506ac993903","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 17:35:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd62cc737e0642bda66431d7bbd0c235 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62858bc22b0a72f71cbf5fcbd8512965f7311593","revisions":{"62858bc22b0a72f71cbf5fcbd8512965f7311593":{"kind":"REWORK","_number":1,"created":"2020-07-23 09:15:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/742615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/742615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/742615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/742615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/742615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/742615/1"}}},"commit":{"parents":[{"commit":"1b0509b52875c480d653870fbea842afe488ae8e","subject":"Merge \"update constraint for validations-libs to new release 1.0.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b0509b52875c480d653870fbea842afe488ae8e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-23 09:15:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-23 09:15:03.000000000","tz":0},"subject":"update constraint for osprofiler to new release 3.3.0","message":"update constraint for osprofiler to new release 3.3.0\n\nChange-Id: Ibfdae9e6e2d85f2964733ff7e94ab2f42da6ebaa\nmeta:version: 3.3.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I3412422241f78b28eded22412300b6b7df09f726\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62858bc22b0a72f71cbf5fcbd8512965f7311593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62858bc22b0a72f71cbf5fcbd8512965f7311593"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
