)]}'
{"id":"openstack%2Frequirements~742629","triplet_id":"openstack%2Frequirements~stable%2Fstein~I0ff7af1b7042163f7ef16d318962b8e39d3ee1ce","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I0ff7af1b7042163f7ef16d318962b8e39d3ee1ce","subject":"update constraint for neutron to new release 14.3.0","status":"MERGED","created":"2020-07-23 09:39:25.000000000","updated":"2020-07-24 13:15:47.000000000","submitted":"2020-07-24 13:15:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742629-1595596547130-1a0e6018","meta_rev_id":"bba5f1b9b104b30e1b050dc080582f9c1a27d059","_number":742629,"virtual_id_number":742629,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 13:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-23 15:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-24 11:29:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-07-24 11:29:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 15:26:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-24 11:29:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-24 13:15:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eadd7a7a2d9dde58ee145c26593068a790e3a62e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-23 09:39:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbbb8e0d1a4a892b3a5d76d493b08ecec0cdb8a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 11:26:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d66fabc94c943be8173b78cf2bc4eea : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47a973224b1d4977b3594352c62cbe0e : SUCCESS in 3m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4a8be40976424b97a0ef39001fb77d39 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ae874b26d0e45998eaecf4a79c3f616 : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d297145c45714a63ad267154b351c5f2 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/689792249b8c4f81a94ffe04b8eb44d3 : SUCCESS in 5m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/789e09ba7f95461ca6a870f5424c4e11 : SUCCESS in 7m 51s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/95bb502fd61448aa9be407bd42b7a47f : SUCCESS in 14m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/52005b95fc9d433297506b87a02f181a : SUCCESS in 8m 28s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0d31f2ed998a4008a07f39caa8a6926d : SUCCESS in 9m 31s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/29c9e16e9fdb47e6bab6fef0ea7426e6 : SUCCESS in 9m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a8d3a6d4fe7644ff9f082db6d931a879 : SUCCESS in 5m 28s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d6664a1642c649239d540f8d52151173 : SUCCESS in 5m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/75c9fd8b6bc64809860714e1f48cd722 : SUCCESS in 11m 36s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ffc837f468b3487b99fdcae492c84c98 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b76e4cc4db9240e6bc4a843ae535e69e : SUCCESS in 6m 40s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/06b4464ca6ab4fb3b3371caee8038008 : SUCCESS in 25m 10s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/955fe762db52480f9c69db06c7e84d69 : SUCCESS in 31m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/605a025b1e9345db8e3e81bec4bcc31a : SUCCESS in 21m 55s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/1498ead1f7cf447a87b4339e892f3c4d : SUCCESS in 25m 14s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/901f1b408ecb4b1a87cebf5e1dcec016 : SUCCESS in 15m 54s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/42644b4580e5443e8ca8e1909a6f56af : SUCCESS in 15m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98846518f9c24ad6a328bb3b9ee3ad3a : SUCCESS in 21m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/652795e928e2441fa24d895896cd5534 : SUCCESS in 8m 11s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/25c5ebe425c944b28a0dd48e5998f7fe : SUCCESS in 9m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6057c56579a541a599dcef57a0f4cd5f : SUCCESS in 1h 12m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1caa29c087d04e5f853e431f366f97c4 : SUCCESS in 1h 36m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e7a6745d66bd4d578169655bc18e34a5 : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7275c6a83b1104c1f4f25e2638c352b2a7c7e6ef","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-23 15:26:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab8a151f11d607531dfaeca69f643c7169b895df","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-24 11:29:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"decb349625c52cc455679601daf833b11858f30f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 11:29:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4176112dada018577af03d1ca3f23ec3d8ab59d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 13:15:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1109e21996524ed78aa78a7dd93f2f47 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d17cfea100524be1b06121308c91a792 : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/09b4171ec42a46f4bfa5b8c79cc499e7 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef447b1290264b77a3ea763c8cc3749a : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08842748d79848869b948569bfd0363b : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9a684f7d3e047dda7326579bd3b9b20 : SUCCESS in 4m 28s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0a147bd7daa84b46ae91b4e429341527 : SUCCESS in 8m 15s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/8d5f8e8fbd74439a830b56bdd1b5e357 : SUCCESS in 9m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/d7f72bcb98334f4c832bf1957e2f6b5e : SUCCESS in 8m 02s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/036a8cba6bee40429d809a96796ebda7 : SUCCESS in 8m 59s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/36dc187d56cb4e758a24f22f2213675f : SUCCESS in 4m 29s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3a2f1fbe36404c788c851257440f761b : SUCCESS in 4m 50s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/34afa87885ea43bf938211feffe260d6 : SUCCESS in 11m 30s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/203e3c175c5a4cb1bee8528bb349676c : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cce04ad6564c4bc89d4846a1dedb9b61 : SUCCESS in 9m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/988a49ff15bb4f3da3bdf0a28e5f3353 : SUCCESS in 23m 33s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/975f6b9cd096495a951bb997ae30b8d8 : SUCCESS in 32m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e1a3db8bffee46048b6152201a56b5de : SUCCESS in 23m 22s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/2e0738576a8746d1a6d9c223591a7ff0 : SUCCESS in 29m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/68d9a31b4df1402bba147ea85c8522c7 : SUCCESS in 13m 22s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/579617eb9cc045f6b8b93fc870b582b0 : SUCCESS in 17m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45390b78655e4522b9dd849dc4f9f87b : SUCCESS in 14m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/57dfa61799494665b963e63127a9a100 : SUCCESS in 9m 13s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/9a7c29ac7ab24d2eb1b6700fd60c2877 : SUCCESS in 10m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3a27b5b278c4972b925cf5517732299 : SUCCESS in 1h 34m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/26f4474b93f7456a9eab92c9ae676a1f : SUCCESS in 1h 40m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3381518ec92b4beba77d144424579118 : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f22ee6a5c371adffb9a95d6128ef32aa6529571d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 13:15:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6589526d39bc3c142f9f8f367b5ab6462a128c19","revisions":{"6589526d39bc3c142f9f8f367b5ab6462a128c19":{"kind":"REWORK","_number":1,"created":"2020-07-23 09:39:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/742629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/742629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/742629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/742629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/742629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/742629/1"}}},"commit":{"parents":[{"commit":"6eb52e0753a4ce5217058bb5d5f935589345bfac","subject":"Merge \"update constraint for castellan to new release 1.2.3\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6eb52e0753a4ce5217058bb5d5f935589345bfac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-23 09:39:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-23 09:39:23.000000000","tz":0},"subject":"update constraint for neutron to new release 14.3.0","message":"update constraint for neutron to new release 14.3.0\n\nChange-Id: I0ff7af1b7042163f7ef16d318962b8e39d3ee1ce\nmeta:version: 14.3.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ifda2a598131009a531d5c4d545380b3bca0a8259\nmeta:release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6589526d39bc3c142f9f8f367b5ab6462a128c19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6589526d39bc3c142f9f8f367b5ab6462a128c19"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
