)]}'
{"id":"openstack%2Frequirements~742906","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I97d02a72fe5935150a206886322ad8175fc059fd","project":"openstack/requirements","branch":"stable/queens","topic":"openstacksdk","hashtags":[],"change_id":"I97d02a72fe5935150a206886322ad8175fc059fd","subject":"Bump openstacksdk to 0.12","status":"ABANDONED","created":"2020-07-24 15:03:15.000000000","updated":"2020-08-03 10:44:15.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f989d5c296c7af12c56a526f85cf110b507ec38b","_number":742906,"virtual_id_number":742906,"owner":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-07-27 11:10:35.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-24 17:42:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"disliked":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-2,"date":"2020-07-29 20:56:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":-1,"date":"2020-07-24 15:24:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":-1,"date":"2020-07-24 15:46:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-07-27 07:58:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2020-07-25 18:03:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-07-25 18:04:04.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 15:06:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-24 15:24:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-24 15:46:25.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-07-24 17:42:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-25 18:03:56.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2020-07-25 18:04:04.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-07-27 07:58:10.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-07-27 11:10:35.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2020-07-29 20:56:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"aee583f0b74292bc1016aeac9ccfe355e3fd6f5c","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-07-24 15:03:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9719f4bec9098e0a98720d2bccab1a401efe27a0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-07-24 15:08:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11ce44173baec7ab6bc008a1f76aa9b0507ca3b8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 15:15:55.000000000","message":"Patch Set 1: Code-Review-1\n\n0.11.4 was the last release on stable/queens. 0.12 is from stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"d8ad7b448e33375ee8bc1e46293ad5fac3de3432","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-07-24 15:23:31.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"244491bf325f8b1d940e99d1acef706c04297873","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 15:24:28.000000000","message":"Patch Set 2: Code-Review-1\n\nSee comment on last patchset.","accounts_in_message":[],"_revision_number":2},{"id":"655dd432e04303e7895f9afa72037980660573cd","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-24 15:46:25.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"224cd7566a70521bb18cd9a238d1e706285e0f88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 17:42:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da985a76e08a40128d4a80d2b5cce3c3 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/763d3e1912d1488a877408d4c214f14e : SUCCESS in 6m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4f316e0d282f40d0a87936ea0a531cb4 : FAILURE in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30fc06a329cd4d17861f41bf736199c8 : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93cb04de1f88498ba75d7d30f3ea51c8 : SUCCESS in 5m 23s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2fa07855c2c046589132d731847d86cd : FAILURE in 8m 51s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/a8ba2c420bf94ee2b4ea161f23a0b946 : FAILURE in 9m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/3d14cc548b7c4b1d9870baa2f3647bd9 : SUCCESS in 10m 38s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b318b6e9a7a0436c82f14fd401b0bea2 : SUCCESS in 10m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c9a00a77768641a2b4e1cd6e433b8865 : SUCCESS in 8m 39s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/a350a7dfa86a4beaa5d8d5cc29eb7ee0 : SUCCESS in 8m 17s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fc5127991fc54418ae092c6f57c29e04 : SUCCESS in 13m 04s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/e97a331ee45e4ccbba8294cba05c7310 : SUCCESS in 11m 44s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a0a7efe286584d639f755ec806dbb8d5 : SUCCESS in 24m 46s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/8f8cf8d977b54fcf8e609b81e437836a : SUCCESS in 25m 41s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/99168254a77a477e802a57d11066a9df : SUCCESS in 23m 09s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/55a605775ab04db18b5b91a33ecb5f4f : SUCCESS in 24m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/334cf72eff0746e0a04ecf626625138e : SUCCESS in 14m 08s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/4168c89a3fe7485797a9d012c9448c8d : SUCCESS in 13m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a361a21a5c24137a3e379cde8d0eba6 : SUCCESS in 19m 10s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/93ee5f41a7fb45fe880f80bccbc0d8e9 : SUCCESS in 10m 01s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/662bb7fc190344b7869580bb8f89f8a2 : SUCCESS in 5m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dea9f9b461ae44e28689be98938d7996 : SUCCESS in 41m 51s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/45b683f4b46644e6a39a28a594c24f1a : SUCCESS in 1h 22m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f5f00b872b1c4fe2b3538620bb8ea4d7 : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":2},{"id":"379cab06c9a1f44e2d505222c2bee72d153154f6","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-07-25 18:03:56.000000000","message":"Patch Set 2: Code-Review+1\n\n@Sean, openstacksdk is supposed to be latest version, no matter what is version of openstack cloud. I think 0.12 is the minimum required for any openstack modules to work at all.","accounts_in_message":[],"_revision_number":2},{"id":"edc77a62e8c684b11daad6dbd6f00bb5793f3163","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-07-27 07:58:10.000000000","message":"Patch Set 2: Code-Review+1\n\nWe are working on tripleo side to os_tempest on stable/queens branch, in order to use openstack ansible modules we need to this version.","accounts_in_message":[],"_revision_number":2},{"id":"1d75a9920cd02735dcaf4e08b478a9d71f649e7a","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2020-07-27 11:10:35.000000000","message":"Patch Set 2:\n\n\u003e We are working on tripleo side to os_tempest on stable/queens\n \u003e branch, in order to use openstack ansible modules we need to this\n \u003e version.\n\nHey folks - there are other ways to get this into os_tempest. IIRC there are overrides that can be applied to change any requirement in upper-constraints.\n\nIs this change in requirement due to something changes in Ansible, or is there functionality in Ansible that needs the updated openstacksdk?","accounts_in_message":[],"_revision_number":2},{"id":"e2b1b7904a85cdee5fcd67fc2ce25a91b97397f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-29 20:37:05.000000000","message":"Patch Set 2:\n\nI don\u0027t think it\u0027s possible to get this because 0.12 is rocky, as sean mentioned","accounts_in_message":[],"_revision_number":2},{"id":"28de3804a213ec803297cd6bbf778c4ba18d687f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-29 20:56:48.000000000","message":"Patch Set 2: Code-Review-2","accounts_in_message":[],"_revision_number":2},{"id":"cdd930e6682a7baf25740dabb2418c20a0e0bc75","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-08-03 10:44:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"61e1dd9e6c268ecfc04c9a629d0af88257f29980","revisions":{"119e5c38083eb8168c6d54a354b7058bb4ff6881":{"kind":"REWORK","_number":1,"created":"2020-07-24 15:03:15.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/06/742906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/742906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/742906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/742906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/742906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/742906/1"}}},"commit":{"parents":[{"commit":"035ca011130f04e6ca6d53eabd473fcb0d9a1aa8","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/035ca011130f04e6ca6d53eabd473fcb0d9a1aa8"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2020-07-24 15:01:00.000000000","tz":120},"committer":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2020-07-24 15:03:12.000000000","tz":120},"subject":"Bump openstacksdk to 0.12","message":"Bump openstacksdk to 0.12\n\nAnsible openstack module requires openstacksdk \u003e\u003d 0.12 to work properly\non train jobs.\n\nChange-Id: I97d02a72fe5935150a206886322ad8175fc059fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/119e5c38083eb8168c6d54a354b7058bb4ff6881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/119e5c38083eb8168c6d54a354b7058bb4ff6881"}]},"branch":"refs/heads/stable/queens"},"61e1dd9e6c268ecfc04c9a629d0af88257f29980":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-24 15:23:31.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/06/742906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/742906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/742906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/742906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/742906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/742906/2"}}},"commit":{"parents":[{"commit":"035ca011130f04e6ca6d53eabd473fcb0d9a1aa8","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/035ca011130f04e6ca6d53eabd473fcb0d9a1aa8"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2020-07-24 15:01:00.000000000","tz":120},"committer":{"name":"Arx Cruz","email":"arxcruz@redhat.com","date":"2020-07-24 15:23:05.000000000","tz":120},"subject":"Bump openstacksdk to 0.12","message":"Bump openstacksdk to 0.12\n\nAnsible openstack module requires openstacksdk \u003e\u003d 0.12 to work properly\non queens jobs.\n\nChange-Id: I97d02a72fe5935150a206886322ad8175fc059fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61e1dd9e6c268ecfc04c9a629d0af88257f29980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61e1dd9e6c268ecfc04c9a629d0af88257f29980"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
