)]}'
{"id":"openstack%2Frequirements~742972","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Iabf57668d9d105ac71dc1c5a76ba71bd2b081844","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"Iabf57668d9d105ac71dc1c5a76ba71bd2b081844","subject":"update constraint for tripleo-common to new release 12.4.1","status":"MERGED","created":"2020-07-24 18:57:03.000000000","updated":"2020-07-25 01:53:40.000000000","submitted":"2020-07-25 01:53:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742972-1595642020785-7a5aba1f","meta_rev_id":"6dc400409d9133ec7703b8cb7e1f2e4989be5762","_number":742972,"virtual_id_number":742972,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 01:53:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-24 18:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-24 22:28:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-24 22:28:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 18:57:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-24 22:28:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-25 01:53:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d84d834eecca2ad006b7950010244e9840970503","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-24 18:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1de23aa01a6d8fd02017b2b3a1da67360be25a6e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-24 18:57:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03169f93754015796d26359515f5e6cd8c4b06df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 20:48:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e954241e0b4740cfb5c74bb33a7f6ce1 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2785f33e55014eb9a6b63c30e85f03bc : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f11990040064a019cefd0ffbdb9bd23 : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/931a9a35a5a14193a42de370c77743fb : SUCCESS in 7m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cbc183b171df48d6ac609412f17c24e7 : SUCCESS in 5m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b203dd1f4454c9f979f839eb51d82ce : SUCCESS in 12m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f45b8b4a9bdf4bdd80000c9c7524ab59 : SUCCESS in 14m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dc610c4cc80142e686e1b2e0e19c14c3 : SUCCESS in 8m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/feffcd101cf74ce5b0c1620c84a7c2d0 : SUCCESS in 13m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/328894abfcc74cad86f3eed40b543043 : SUCCESS in 12m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/514335104ca54c5fb5c000c0f1722795 : SUCCESS in 35m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f5e7a88fc2cf4147b1e46d21817aa566 : SUCCESS in 34m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b35239cdae74e259e85d31b0f4032f2 : SUCCESS in 25m 25s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/be20794e7187481e8769620741ab43b7 : SUCCESS in 18m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ae8603b2e3e4cc48ea98fe8423187ce : FAILURE in 30m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b5fe6508fa3f4582a23710cc141aaa3b : SUCCESS in 7m 03s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ba143f25f025423fbda57f1176e86276 : SUCCESS in 8m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d905b716d4b4dccb2068f04365d44c4 : SUCCESS in 9m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9171cc9a58e84035ac8901c0ec4bf37b : SUCCESS in 16m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b838b567890040a084c3a32362b1ecdc : SUCCESS in 27m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f66822c371f84a3db124a804eccfda00 : SUCCESS in 51m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b6c7cec6e1424168a304dcff20f55b21 : SUCCESS in 1h 48m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/627112496ce1476c94aef82253da6595 : SUCCESS in 1h 32m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/40e7c00593564b4ba467ec256901e095 : SUCCESS in 1h 40m 47s","accounts_in_message":[],"_revision_number":1},{"id":"54c40b2ea3d0fa2c1e75444f19535b095715b3d7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-24 22:28:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f715ca57d8a749d9f524adb9e4cbf4c24cca9671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 00:01:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/569bc3325ee04ec3a612a77973789ad5 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc4b48994077494f94479ff13c8f2b65 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f26ae437d3f84af2b69ecb50be980a71 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cbc023973e684d11ab8a53f7a153e78a : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b12e72bb87ec4ebea41a74f5909a2649 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a27b9fdfcd474d6e936b5006e79bb22b : SUCCESS in 7m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/151777c4db7e4111a342813f4841abcf : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b292bba6eaaa468682c89dd0468ca3a1 : SUCCESS in 5m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4e10f196219a4c8eb188722f66848c6f : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0a8a6b5b92b0421a879bcf1e137afb8b : SUCCESS in 7m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5deb18c869cd4a53ab30d256d301c78f : SUCCESS in 13m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6dd430f3bf3b4fa1b5c806dff53eeb55 : SUCCESS in 24m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a800bb081c314c27af8dce58e27d00a4 : SUCCESS in 13m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/38f5bdd7afc346d2bd01f79a31a28b63 : SUCCESS in 13m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e374c110f1f4918811183d393110024 : SUCCESS in 17m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/89f2f14cbf1a4968a0616d567c6d985c : SUCCESS in 5m 29s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/bf80b4ab12914a2bbf6381fc2f873524 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/effce7c9d42d47ceab614c4c2aad92e3 : SUCCESS in 6m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7e164f6475b54069820897a5110e7045 : SUCCESS in 11m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e1289950cb97498fbc845142ccbb95d2 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbbc17395e1644109ac698d7dfdfbf00 : SUCCESS in 50m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aea2944b71f04bd4815ecccee0cb3a33 : SUCCESS in 1h 26m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/172930877a774399b195daa07899cd3f : SUCCESS in 1h 13m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/51b8da70761943eebc4f494805ffa1f2 : SUCCESS in 1h 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ef5b677ee7819ef85f8a52ae153d5e8fc1132045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 00:01:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd54dddf101129431fe7581bc07d81fdf76bcac2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 01:53:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"72691590e114b1e1bbb53f6b13e53a81e3930ce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 01:53:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/522c295c2ae4448da2f455a403b9e39c : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14e234141bdf445f95a6bc7727c7f611 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/957f3451a0d84c6581376124f97993bf : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30f6201c29ae4e0bab703f5ef6f43637 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8f3581e5e12048ba8fe8d57d50524e59 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3f6ceb7f9c52477a8d10df35046cab6b : SUCCESS in 8m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b268ed59bfd240c49079f5ef5be5c884 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/516fda9d87fd4490898fad697d22f298 : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8bf3162be47a40f997473da832e67ace : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/308668536c5d449d80f3174a796b4ded : SUCCESS in 6m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2a68814fbdba41c984b7beb6e5600406 : SUCCESS in 20m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/674d3bf2671f4fe6a89b7b1357bb620d : SUCCESS in 27m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c4f378ed1f2e43519b1c6049522d155d : SUCCESS in 14m 06s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7539e548fdac47428a957bb14cfa5d14 : SUCCESS in 12m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/318954172e3d49c1b81ae2960489ec03 : SUCCESS in 17m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4938bc44e760416b85d924980427fd1c : SUCCESS in 6m 11s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/03adc857f5c94a7283db771903a5f185 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5c07ed349a64ebc8f3dbe816856b353 : SUCCESS in 7m 51s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a9ced44ae11e40379e81a7a98d6cbf09 : SUCCESS in 10m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fddab4fbc8814d4e86fcba6560a250a1 : SUCCESS in 9m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01d672f8d0a24bac89fadef02337fb70 : SUCCESS in 51m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/82fdeeeca2f74f2eb97d0796f370d344 : SUCCESS in 1h 22m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4342b9fb18704c5e998141c992694b6c : SUCCESS in 1h 12m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/49d91a81b3ec4bee9ca5b32d5e4f72f3 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4a237b97c4fcd2dbdf77e84ba9922addcdfeae3","revisions":{"b4a237b97c4fcd2dbdf77e84ba9922addcdfeae3":{"kind":"REWORK","_number":1,"created":"2020-07-24 18:57:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/742972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/742972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/742972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/742972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/742972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/742972/1"}}},"commit":{"parents":[{"commit":"6bf9795cef6fdb7a258cc48780db83c231468d63","subject":"Merge \"update constraint for castellan to new release 3.0.2\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bf9795cef6fdb7a258cc48780db83c231468d63"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-24 18:56:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-24 18:56:10.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 12.4.1","message":"update constraint for tripleo-common to new release 12.4.1\n\nChange-Id: Iabf57668d9d105ac71dc1c5a76ba71bd2b081844\nmeta:version: 12.4.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Change-Id: I79f061f57f04c155cf99ba80d8d96c935521e75d\nmeta:release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4a237b97c4fcd2dbdf77e84ba9922addcdfeae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4a237b97c4fcd2dbdf77e84ba9922addcdfeae3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
