)]}'
{"id":"openstack%2Frequirements~743397","triplet_id":"openstack%2Frequirements~master~I1b3619e98a23194df825ce94e969ac6aef4223e9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1b3619e98a23194df825ce94e969ac6aef4223e9","subject":"update constraint for oslo.config to new release 8.3.1","status":"MERGED","created":"2020-07-27 20:22:14.000000000","updated":"2020-07-28 02:14:11.000000000","submitted":"2020-07-28 02:14:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743397-1595902451682-6bbcf8b7","meta_rev_id":"5b2c4789ee0979f913bd9aef6fd5f683a755d1d4","_number":743397,"virtual_id_number":743397,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 02:14:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-28 00:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-07-28 00:47:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 00:47:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-28 02:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb2eaaa3b96568483191321676a8fc2b789a7500","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-27 20:22:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68472046d46f5184664e5431c10d4ec17c4ce3a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:26:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"eb4a1b51b31f1891188a7e3991026b50c13163a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 21:46:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/423c73d668ea45a1b05a839679ecdbc2 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fadf4876963947249429244915278301 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c06775d34f554f888e1a28fbae5594ec : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9aad6c8bd6e14eb08cce3a38053778f3 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b8a9ac8e92646e480c77277ea7331fb : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9452dc6d8da54e6c92d8a7d694f350fd : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/88c8481903444cddb2cf099610b311bb : SUCCESS in 7m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/76a8fd7a236d41d0b6f215c9402d58c6 : SUCCESS in 10m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4ea270208bbf491aa44c43360aa542fc : SUCCESS in 8m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fadac789b0334c829584aa4bde6016dc : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/85be5aa0e0674a3ebb671fb7e767e85c : SUCCESS in 7m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c5a3a29ceba40eba7ae88cf2ed05191 : SUCCESS in 7m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a11a0eeeab34a739a4d9bd6a94def41 : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dae06d41170c4a44b15428f190aaf16a : SUCCESS in 15m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9039562a3ec24bb0a9866460ac10048d : SUCCESS in 28m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3835fc5dcd5540bdadb87eff7c4122f8 : SUCCESS in 14m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7727cd1d6cc34e6b8eec75020d704313 : SUCCESS in 23m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d9eea6a619e443f1b8533a88fd271312 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86f6d9ed093446eb9ef0f6a183c1f907 : SUCCESS in 7m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9b7e93e2aee44b1081277a910e846d8e : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4073a3e11bbd40e78c5c469bd98c2e32 : SUCCESS in 52m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/701ff63b3d904515b619afb296d30157 : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":1},{"id":"087296671b3d12b33709a3c1479c852b535c2a99","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-28 00:47:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"07bee44b9b953cecffc69968d6f17c4f23957ea2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 00:47:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c3e326b4445c80cdd611370d1a36cfb5ff50c01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 02:14:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0fb4a770973cb84395e50a4ab7995c990ebff5f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 02:14:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0a074e1c49e4db68004b9fb4cfcef2b : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e189d4c562428ab4147ca6b66e4b49 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e76d17011bd740dea1d757a2e39997b4 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14b33936aaf641e29a76c567424fd9a5 : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df896ce35f534c57abc8b78846610234 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/675ca505440d4ec5a9469a5e57eebbd1 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15ccf6667d474a629386aba7593f4119 : SUCCESS in 6m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1f3861de846343639ed475cb0efc2fea : SUCCESS in 10m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5403656571c84a04b954dfe0ea9b94aa : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b5637d1a1c5f483aa89b9dfa645f8ceb : SUCCESS in 8m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bb85d16597594f6e95cc8274a97093bd : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0cc1d28bd9404ab9b11bb9575e8678d4 : SUCCESS in 6m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/148fa6479dba46da975e0afedd085953 : SUCCESS in 7m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7ded7bea702c45328a7d277b8dc2737a : SUCCESS in 29m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/71fc31e546cf44bbb25f440495715b39 : SUCCESS in 23m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/221ff0365b8e4b4780fe47064cb0eb2c : SUCCESS in 11m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e03dfcb5efb54b5484efe38f25e765d8 : SUCCESS in 17m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/62faddfea9e545eaaa62ee94fcccbc7a : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22d938736096494fa2cccb5f66ad873d : SUCCESS in 5m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/447193f569564aa387956da36e929b09 : SUCCESS in 11m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6022b3a6055c4cbf820be49e971578bd : SUCCESS in 8m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd62d3271b6e43ffa502af744ce82c46 : SUCCESS in 1h 02m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/899b103b20014d19af51a505f85b39df : SUCCESS in 1h 18m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12311b7bb456f1cc115501277d7699668d5b4a6c","revisions":{"12311b7bb456f1cc115501277d7699668d5b4a6c":{"kind":"REWORK","_number":1,"created":"2020-07-27 20:22:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/743397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/743397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/743397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/743397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/743397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/743397/1"}}},"commit":{"parents":[{"commit":"4eba7e628b0290047fbd5df104dbc51284c50e4f","subject":"Merge \"update constraint for XStatic-moment to new release 2.8.4.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4eba7e628b0290047fbd5df104dbc51284c50e4f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 20:22:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 20:22:12.000000000","tz":0},"subject":"update constraint for oslo.config to new release 8.3.1","message":"update constraint for oslo.config to new release 8.3.1\n\nChange-Id: I1b3619e98a23194df825ce94e969ac6aef4223e9\nmeta:version: 8.3.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I405c7ae3f41261b16e8a2408f04626ead2e28076\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12311b7bb456f1cc115501277d7699668d5b4a6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12311b7bb456f1cc115501277d7699668d5b4a6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
