)]}'
{"id":"openstack%2Frequirements~743398","triplet_id":"openstack%2Frequirements~master~I3efa573166d5e67d31b7a4ed501d64ee9c752c9e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3efa573166d5e67d31b7a4ed501d64ee9c752c9e","subject":"update constraint for python-troveclient to new release 5.0.0","status":"MERGED","created":"2020-07-27 20:22:25.000000000","updated":"2020-07-28 02:14:13.000000000","submitted":"2020-07-28 02:14:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743398-1595902453514-46e369dc","meta_rev_id":"76e6fe3fda7b770f2d631a56950f4aab931c1357","_number":743398,"virtual_id_number":743398,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 02:14:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-28 00:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-07-28 00:47:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 00:47:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-28 02:14:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f599b059cad3ef6ddd8f60cc863683b0a2efd40c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-27 20:22:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a53a5848e6ba538d8d63070d5d291387f1306a4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:26:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"8ceadd9043aab5ddcbb64a730f21fb84bde959ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 22:04:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d21155e4a14e328d7f8d935f6be9cc : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05e8c4f620a74d029764e97a5e2d152a : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/835648f4276f4b66959c2bbc3136d348 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a05cca8ed922489ab4bf29acbecbcf0b : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c99dfbc14b184f119879ddaade08be79 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5921dc6d4904a68980c45d6a5f8f766 : SUCCESS in 5m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a41dbf7ccc294d3ea95c1a96f4755483 : SUCCESS in 8m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2ee2319ecc9f4a9495d36632d503ccc8 : SUCCESS in 9m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5b498d42ff2b4704a57288519f5d7df1 : SUCCESS in 11m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3dc9836a9dc64b4c8ed7bf050dbcdf20 : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/512c0a02628241cf84541ad8a07fdd20 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1b491253350466fb57be193d6773f54 : SUCCESS in 6m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/73bcdb5e94cc48b59047fb1457152da3 : SUCCESS in 8m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/32722d5d8cf646f09bd74b39af077abb : SUCCESS in 17m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a5bac548dd1d4ae9afd007ed2076d2bb : SUCCESS in 26m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f5da2d85f5eb4a6584cd61dba02e906d : SUCCESS in 14m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/54bc7214865b4f29b1df739c0a2a0938 : SUCCESS in 16m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0b355bcb508c47b1abb9c9fbbbe1dae7 : SUCCESS in 7m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d8e5e86af0249ef866f5c6d94547087 : SUCCESS in 7m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f49ae43c6c8342869cabf662bacf8d37 : SUCCESS in 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd70f35d73b249ff84d8766a733239b0 : SUCCESS in 49m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33b590bd9efb45d98a3366a029a2e208 : SUCCESS in 1h 26m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6e7a02ea17f8ae8a24ebb364f361a93de30e1497","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-28 00:47:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90d3210b9d286d746dbdfeac82d5909c65365a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 00:47:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e9ecdd6e229fe1f987fc41b455ec2be6aac05f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 02:14:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f9d766a5d31d06a62ceab27ff4f9c91a4ab863a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 02:14:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c9182adcc046a593dd7e80cf53e1a7 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/507b3223dea8451796ce5be5607e7f2c : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/538e6e3002f141bcbcd70cb2becd556c : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21b9981272c94b09b853eefab453dffa : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa48b2812a124b22b9edf47e4726eb39 : SUCCESS in 3m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/41a78f67b72343d9bff3ad261fa36b74 : SUCCESS in 5m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5115840157d743aba347287acea9c4d5 : SUCCESS in 8m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a7245628ecad4f4a950a0853caf8067f : SUCCESS in 8m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/92d9351096ce488f914284f742fe662a : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/78a8651de3294a8e8fa92d8b2dd20f16 : SUCCESS in 4m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b70769aa1e2a4c93838c5c3b64d8020f : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/294d950435154fed9f2fe18566d7a7d2 : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/58fbfcfeb7c14a71b691b3596c1612d2 : SUCCESS in 6m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a28311001d984f3ba58895e6a8c16433 : SUCCESS in 23m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e475f65e4c1947438744e75642a5e253 : SUCCESS in 17m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b8b4f31411f548eda210bb903673a5df : SUCCESS in 12m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9fb816bc6d064d6ab5b3c1233dec0f62 : SUCCESS in 18m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/65545924abed4b24aa8a36011248be35 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65ca620c3b88415caecf668ffc5dbdbb : SUCCESS in 8m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d61635e2027b439ab34f46610f4d5f31 : SUCCESS in 10m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c166abdec8d24df68acf50bc75808059 : SUCCESS in 9m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16f028e89f664aeab853e7b8f243802b : SUCCESS in 49m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b7accd19bad488c8da9a1f76d460e41 : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7852aceccb7e99c84cd7795802f66489df2e2241","revisions":{"7852aceccb7e99c84cd7795802f66489df2e2241":{"kind":"REWORK","_number":1,"created":"2020-07-27 20:22:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/743398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/743398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/743398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/743398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/743398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/743398/1"}}},"commit":{"parents":[{"commit":"4eba7e628b0290047fbd5df104dbc51284c50e4f","subject":"Merge \"update constraint for XStatic-moment to new release 2.8.4.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4eba7e628b0290047fbd5df104dbc51284c50e4f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 20:22:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 20:22:19.000000000","tz":0},"subject":"update constraint for python-troveclient to new release 5.0.0","message":"update constraint for python-troveclient to new release 5.0.0\n\nChange-Id: I3efa573166d5e67d31b7a4ed501d64ee9c752c9e\nmeta:version: 5.0.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Commit: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Change-Id: I3ecb4927f69874f167276651718ff3aa878f6755\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7852aceccb7e99c84cd7795802f66489df2e2241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7852aceccb7e99c84cd7795802f66489df2e2241"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
