)]}'
{"id":"openstack%2Frequirements~743401","triplet_id":"openstack%2Frequirements~master~I9e274eeb55a0455a5f15740a192791babd38b7d5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9e274eeb55a0455a5f15740a192791babd38b7d5","subject":"update constraint for oslo.policy to new release 3.3.2","status":"MERGED","created":"2020-07-27 20:23:57.000000000","updated":"2020-07-28 02:16:36.000000000","submitted":"2020-07-28 02:14:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743401-1595902455602-d52aeaad","meta_rev_id":"7513c72ef4259fbc3a4bf47e1774a209cb9b360a","_number":743401,"virtual_id_number":743401,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 02:14:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-07-28 00:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-07-28 00:47:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 00:47:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-28 02:14:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6fedc3861587abb771e1b42448bb2a40bc55d86a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-27 20:23:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2807b78c95223cde80527dafd34f6ac80288ed8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 20:26:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"adaad169c112998c3208d864b141f68bc9c55faa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 21:48:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffeb262751b94b6298cc72dc063ad59d : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ec917becd984b1bb144d4d06f9e5483 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e03e3b34cc3b420a8beb2ad9c0dfe7a9 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b88267b6899547c992befa59ddfc017a : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e45d8198b4e54d7baef2475e87951621 : SUCCESS in 7m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32050211efa64e33884f32e95050a798 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b4ef2e0c2cd4433b3a516d2abad3e5c : SUCCESS in 7m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8254811d4af04c999988d715192705c1 : SUCCESS in 9m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/acc8573fb0354487b61c8815799b489e : SUCCESS in 9m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/38e9d196ac3841bcaf964fba67d28a1c : SUCCESS in 4m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/83650ec765e5492bad0b02f3771ae429 : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ec4e25e3ace84c2faf87ac3fa8a16a00 : SUCCESS in 6m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/53d83a4d1fa84328beb2e4c753c51eda : SUCCESS in 9m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/404feefc31924ab98e7b9bb88961083a : SUCCESS in 15m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5c2a96a4d764413ab0ed742d9e06fa7d : SUCCESS in 26m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cc5dac83b70e41acbb6076e760c1ef2e : SUCCESS in 12m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5fa4ac6db7f040eebca33688a580cdab : SUCCESS in 16m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ffb0b6053f824fb2bb210ac99b0af2c9 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe3b4d40102d4545a0e2acb36af8c8a7 : SUCCESS in 6m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/54eb09e19ad9493eac02d25e94733f24 : SUCCESS in 9m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b492c41703a74706b4dda211bd035210 : SUCCESS in 55m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/338c86c3108948ea97ab846259b70b1f : SUCCESS in 1h 09m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7820a933fb63c0e400419ab61f94f5c1a38cad4d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-28 00:47:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f4f5f9a2c53df6fd3a0ee10c5319f1d3621022e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 00:47:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"542459faaf19575fc8b0c3ad41e5296b0ea6f420","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 02:14:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa0ee5aeed5239f56759a97c5e5c8b202cca14bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 02:14:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7d472057cd045e18f3aa2b6ba197c54 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c6c87a484a4ee1a742b49ca02c0150 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b40f0389a974d81b8351afa1a0aaa3e : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/834e023eb0174aa6b54cad31ac771852 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29063bb7415b4bd7ada75940497ca5bc : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72ce3864b1944bb393a15e15a8614cb5 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/26bf8fee5d2d4ecb8cb92873f9214530 : SUCCESS in 5m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ed124d9ff584d85b70ecf20389ddc03 : SUCCESS in 9m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c910bcfc82314fc78f3455692a13d5dd : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6a48f9c9682a447c83595be50b25e809 : SUCCESS in 4m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/aa5979a242ae40b59bf24fa705a6b5b0 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c251765a6d04a0d976d896db3f96350 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3a953d13eb1443278ef54b73aa618e4e : SUCCESS in 6m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/25884bef165149c9929e996935189eb8 : SUCCESS in 13m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3de7eec7c6d64d99bb68909a3f075dd7 : SUCCESS in 24m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/abb9783943a74db3a1b8d5bd39854163 : SUCCESS in 13m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a371abb01524871a361e1b460d7185d : SUCCESS in 22m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6f5966a0cc034b698202b7f486c480ba : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1dcb16b16ed14ef5856be7527581678f : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e31430f39e2b4d17a20f7b775b67a74c : SUCCESS in 9m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f95cc1750aec4ff7b15bf522d145bb99 : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/955cef4e299149a8bc155bc7b90311f3 : SUCCESS in 46m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a22ef34d9c2c478ab2dc7b9a87f0d947 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d4f3b3949c48f32f341205e0f369be8225dd27d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 02:16:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e848fd845474a7cb7282d7853b64931 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"deacd387487a147e9bde0aa2414d6a589db3d0ca","revisions":{"deacd387487a147e9bde0aa2414d6a589db3d0ca":{"kind":"REWORK","_number":1,"created":"2020-07-27 20:23:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/743401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/743401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/743401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/743401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/743401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/743401/1"}}},"commit":{"parents":[{"commit":"4eba7e628b0290047fbd5df104dbc51284c50e4f","subject":"Merge \"update constraint for XStatic-moment to new release 2.8.4.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4eba7e628b0290047fbd5df104dbc51284c50e4f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 20:23:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-27 20:23:55.000000000","tz":0},"subject":"update constraint for oslo.policy to new release 3.3.2","message":"update constraint for oslo.policy to new release 3.3.2\n\nChange-Id: I9e274eeb55a0455a5f15740a192791babd38b7d5\nmeta:version: 3.3.2\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I405c7ae3f41261b16e8a2408f04626ead2e28076\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/deacd387487a147e9bde0aa2414d6a589db3d0ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/deacd387487a147e9bde0aa2414d6a589db3d0ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
