)]}'
{"id":"openstack%2Frequirements~743463","triplet_id":"openstack%2Frequirements~master~I1ad04d2a9e2aa35cf9d90e00d4137c34a62fb6d5","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I1ad04d2a9e2aa35cf9d90e00d4137c34a62fb6d5","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2020-07-28 06:30:10.000000000","updated":"2020-08-05 15:36:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0adebd15e88d7ad0b2fcb622d0e10b83133beeb5","_number":743463,"virtual_id_number":743463,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-01 22:12:38.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-08-01 15:37:11.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-31 22:47:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 22:47:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-01 15:37:11.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-01 22:12:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"27031830e73dd54efad1e687fc0bdee477ac50d1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-28 06:30:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3aeea95c96d7fb15cba896670971f37250340fd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 06:38:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"81ccf3e725d3df692444bd3fe26eb5a993f25f25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 08:52:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea169931f97f4b1ba82ee09bb17f3735 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d478c1d7df9346be8561af9f9a014f51 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/077c1bd5e8d443cab324a1ab9426e268 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daf4fcc76981425ca8f0ed65521d1416 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/71db996dee7e44feafe51ef71fb6a0ab : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c764a955cc14b17a7b5e75e10eff592 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6b2eb5c4cbb4fd59a65fd950b725b7d : FAILURE in 5m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c4e44b9575f141e6b963629793dd3386 : FAILURE in 4m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4748d56c01ed4046848ef143a613ef51 : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5939ed22ac0e4ad1ad3eed6777ad76a7 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/65c8bcafa64d41c6abdb54dcdb3a8230 : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c912ed43ad9473f818517133511c54a : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aaec561cad8f4d28985562977944882a : SUCCESS in 8m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4af2c93845bb4f9697d159121bbee56f : FAILURE in 22m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f3449e5ce95642e4af132a67cbb55441 : SUCCESS in 27m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a6f7ce5315a84901aa518f9b60e9d57a : FAILURE in 14m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2819ce0f32c445e58bb944bd96e3d672 : SUCCESS in 21m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/adf67c0d1f614c95b2b464560a1350c9 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e20f4b1d17084bfb91311185cbe032be : SUCCESS in 6m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d36f6075478040a7a4d4ca7dbba13b3f : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44846774e654441d9c02ce6902e484d6 : SUCCESS in 53m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/654fa686b9e34556a1c8441953dcd852 : SUCCESS in 1h 29m 22s","accounts_in_message":[],"_revision_number":1},{"id":"21e83a421c5baf05633926dc8a62e62a689f7adc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-29 06:23:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3296b2ff1dbde4fc15123b32e191eca127f8f09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 07:06:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"cab134a0081585fa333f1be389fcd0aeedbadeb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 09:35:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e05062ac794f3eb6dbd28b2a32a4b7 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a5377e2a184faabbeb2bd15e7365c3 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f59604d6425409c844db3ac4e8c0986 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7400c7b7a1ca4f64ae6d5f283eb83a36 : SUCCESS in 4m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/deaaebf118ce4eae9cc1d74f032f789d : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c2d6feabbe9458cb898be641c30c7c1 : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4baa6a95596d413f82f8198b541b1024 : FAILURE in 5m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/135375d4b5824de4a5aed45daefb2a59 : FAILURE in 4m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/30b02bc889bd480fba65828d56b2a027 : SUCCESS in 8m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/649e3b948e074b2881ec4cccd3dcae58 : SUCCESS in 6m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f36eae024ef2428a8403be7bee324e01 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13b22c13577640c6a1fdbe968cb1f675 : SUCCESS in 6m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4f7f22de484b4b4e98024648b7f34fe5 : SUCCESS in 7m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ec71293b267a4417a26932627ec3f781 : FAILURE in 14m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b1cf1846d50447bfac505a0ba90be42d : SUCCESS in 29m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ef1a31e1f2cd4ac397e9fb4628924570 : FAILURE in 12m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0d54ec3b2dff46b1a64ee85eca92b714 : SUCCESS in 16m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/adbdf9b4598f4626b557fa1757b08b3e : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf1fbf323e4142a1aebd0da895563d9b : SUCCESS in 6m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7d74c94c0d1446c1809c23915c6ad0be : SUCCESS in 15m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/944ca0fb17b94506be819bc9ad7a4330 : SUCCESS in 46m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/794915054a7448cd8101f7af971cde4b : SUCCESS in 1h 27m 22s","accounts_in_message":[],"_revision_number":2},{"id":"358ff862bd7a83ae22827f860e39387b187ae75f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 10:15:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1319a3c9b643612c18b3de3791bc24369d646257","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 10:19:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"9e3aa9d9ab31246bbae08e6c1d6a69e1138ad923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 13:08:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9369b75c65d5400f8f6462932baa7c42 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c00f08814922435ea63d31d2d308bd40 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1827383cc70c43fa949a7b4f0fbbffe3 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df7425e1656b433d995abeebc92b0c37 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32b5dcda55b24f88a753f2599829a9ec : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61595aa9e36e4bceb33bdd9500296ae8 : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cfb908c5833b4597bd4ae834b027c2d0 : FAILURE in 5m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/91b3087023904f7e916d715491eee751 : FAILURE in 5m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/55c19a7febf34a8ea40a449cff43efa8 : SUCCESS in 8m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6993dda9a1fa460b8b4eec566b494399 : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b913c9f6af3a4be8975f49ee267350db : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/295230ae5bdf490f9cd3fe41d2366c9b : SUCCESS in 6m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a92cdbfa604444779d063711dfa8dcf0 : SUCCESS in 7m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e76ef6b3334484cb6b6b02e767862e5 : FAILURE in 14m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d48705e5cf2345c9b9db62522cfe0a50 : SUCCESS in 36m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b88d75af884143c6bcf79714ceaaf8b5 : FAILURE in 14m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4a6f705486e24dba9ef4500fabdae342 : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6d22dfb183f945baae86312f777bc33a : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9eedfef7d233431f8df38d28dda98d00 : SUCCESS in 7m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f56d54ae267e46c1811fb5bb06f3ae8a : SUCCESS in 8m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4167a3367f3942e4b8f8f1ae2db922c1 : SUCCESS in 38m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/947a4bb816344be799cd0627155c3980 : SUCCESS in 1h 22m 23s","accounts_in_message":[],"_revision_number":2},{"id":"97da953d62d3ed1c11aeb9dd7f34ad6254078a33","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 15:12:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"856b05529ee99be3304638103c48efca95ae6a9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:27:21.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"370b080439d4955383848e4efc060307c566483e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 22:10:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0d87c169b0403c9e8b701a42b07c75 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87eaa57849a344fbb925c4a6fbbc84c5 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f8774322c474a54bdbeb6e83334c6d6 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b3f7f3cc3b0425090f32bdb7ac68049 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e42d13fcff0485abe12d737bb96cb16 : SUCCESS in 5m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92e6fe87d1f943c9bb27f00a5c7ae5c8 : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28813eab3b3d4edcba8dff6bfd46fd19 : FAILURE in 5m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e044e330aa764b938e3b87dce7f0f767 : FAILURE in 5m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/922eebc13b7a466a8d0adf9161ec6ed3 : SUCCESS in 8m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/370e7788a8cb4cbaad6de0a1ef825f9d : SUCCESS in 5m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/73f2474dd06c4dcb9419d7c9c2805301 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98bc3c7f890b4df4a3b3f0c77164c223 : SUCCESS in 7m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b8a184699f7b46b5969276102dcea2c9 : SUCCESS in 7m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cb1b2c29473e499195af52bb649ad3a3 : FAILURE in 15m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/024087c9e2354c25b0196f2d0bc794a4 : SUCCESS in 31m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ed4848c6534c436bb5debd671f6ef7c8 : FAILURE in 14m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/66eebddf06b54c46a0228a6d2e01116a : FAILURE in 19m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/adc0122bd51e43c98c0b3ab981f2085a : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/614407b36df24d60ae01067419a54375 : SUCCESS in 8m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/df5e255ad5044a9183cfb7a903762876 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1e3887ca7234f20acc08ae44cc452dc : SUCCESS in 50m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcb60511da5a47fca6fc0230270ff3ce : SUCCESS in 1h 22m 39s","accounts_in_message":[],"_revision_number":2},{"id":"bbda2572798681fb99c3dbc44f12ea691ac17dd5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-30 06:25:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2bb26a6b0978fb83b310b859ebeb50453fc08bc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 07:07:34.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"052808c922b4b8c1afa680cbca7bf7c28db83b25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 10:20:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55c97ef6e6d49348bdddb4c8489cf20 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1d1155b09d4709a73f0267a0d24fa5 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b67e6a433395437b9c869e48746ced44 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71901b40879c4a9682e48ce9fbc298de : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c22e13b435ad40e99bd3c9120425279c : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d78b856238c042888ed4585d0c810694 : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8a719ef9f8ca4dbab0f8769e92af212d : FAILURE in 5m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/530b5660c75e4b99905e04ea1ef88fce : FAILURE in 5m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aedb5acccbbe4f2abf1d6cb82b5e2dd2 : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/505d33cdc85d42f3910423c63db65b2a : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d80a8ab92e9c415fbcc40924576d1b15 : SUCCESS in 6m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50da46968687484e8ba1f430297e5323 : SUCCESS in 6m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6faeb7f2825d44b19816c76175b54911 : SUCCESS in 10m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/534a017086c6460eb94395fde80b057e : FAILURE in 12m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/219fb19a52ab48cca41e248868835ef2 : SUCCESS in 26m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a57283f219c446aaaba94b771f3ba4a7 : FAILURE in 20m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38dda8f70b304cefbd599e4f39bd2ab3 : SUCCESS in 18m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c3d1664bf1ac499a8fd4500100bcb3c9 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e8c2d206e75849058a9a078f69a69797 : SUCCESS in 5m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/63d1a41d6a6e4b4ebd85922449a62d37 : SUCCESS in 9m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bcd0170334b48c2b8fde2f7f220d4f3 : SUCCESS in 1h 04m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a05f785ec5b4622ae4a125db2c6747d : SUCCESS in 1h 42m 08s","accounts_in_message":[],"_revision_number":3},{"id":"abf88ebf2b2309c47fb3a0c1b684577cbaee67ed","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-31 06:32:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f44c14cc3b84e28aa54e00535cf14040dd0f070e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 06:55:31.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8ea259a0487344a9b2749e5349e8decc : FAILURE in 18m 14s","accounts_in_message":[],"_revision_number":4},{"id":"d7790162bc241c26b517199187b5f99cdee366b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 09:08:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b434f9844cb543289d36fcd4f53ebddd : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00c36cd6b9244eb2839e665a23c7e3f2 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d293f6e8af644789b1b4947caf6caa01 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/991001515a994629b4073149fe23b376 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f61cc9f7f9304e1ebd322b9bd0bc3c4e : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7179746aa834e68b776b66845fde624 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6bb5695de3094fc0b75a5d9b5e17e828 : FAILURE in 5m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/050285520ccc4f9bb74babfff3180153 : FAILURE in 5m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ca0bbf67d57447fdb25bfdc03b02b523 : SUCCESS in 9m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0c7469c388704f25bd316c218bb3dc50 : SUCCESS in 6m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e4b74b476c264df98cce029bef4741aa : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60d06bf8ff724cf7ab89c44c9b8822fa : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/92208e5e9cdc40139196026b0f8b431a : SUCCESS in 7m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a100f6f9a9db498d85e9fa79a5babc7a : TIMED_OUT in 1h 01m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e140b9f4ba1747e7b0c0f36081f41a72 : SUCCESS in 25m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8dc3e07ff7a642988bb3f50dee5a1fab : FAILURE in 12m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/264ba8395e344cffb959d178e07e5259 : SUCCESS in 17m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/43897bd181a84a2b99e5c4bbe47bb5df : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/952683fdc5724905b2c4c7157ab8ba88 : SUCCESS in 6m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4a27ffd2ac864681b9d7f1a5cdad096d : SUCCESS in 8m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/377c71ef490e4cd288f2d81a2f76c6d5 : SUCCESS in 57m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d948f558b3d346c1a48a8b5f9d550714 : SUCCESS in 1h 10m 30s","accounts_in_message":[],"_revision_number":4},{"id":"f028667c56620c872df33c71c8dfd6bd7217d9ae","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-31 09:29:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"336c1c54d90f1dfe47ca37d3b9f5c7230d4da8f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 09:48:05.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/db953adc950648588df85edf74533c5e : FAILURE in 13m 04s","accounts_in_message":[],"_revision_number":5},{"id":"72bb50e77541d9b024406f0521e5505cfc48456f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 11:45:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6474bc0936da4e3b95641e7e4e2f3872 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb6415b1d95413cbf84975c3f9caceb : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c19f3df3ab54aa78d4f4efffa55c187 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1df3eb89c70d4e50abe4bed30e8f4485 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7f93f6e521641209387aac38798f7bf : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa134862ecbc406698123d32161157a8 : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/006fdda2c0434e3285c1c65aa9649629 : FAILURE in 8m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae3f9bd7428048f28c9e36dc538908a7 : FAILURE in 8m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2ca971bf420947ef8e9b39a0b9a3be19 : SUCCESS in 10m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/25947a4a5e094e17b3268f8970f2a55a : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/27e6d784a25b4f02b15c8e9b09a4c830 : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d2457fdc79b494ca67e9df290da12f6 : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bae16a69a9dc4864a37abdfe18960477 : SUCCESS in 7m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d2b267e9bba043858c37cb35dc570606 : FAILURE in 23m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f3977efbd21140d890fefef3fc80c1b9 : SUCCESS in 28m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7301585d923d43bd986278a00ce3dc20 : FAILURE in 15m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3cf6d5b4b04431380e1ef5425a38dd6 : SUCCESS in 13m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/acc176854d1c4fa988bf0bc264ea400a : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f162c54a89b841f3a8f75778fee2c589 : SUCCESS in 6m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/78e14c99e6ce4ca691c16915a9d8ded5 : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/585af75e6a404d4fa3eb87c03102e2c2 : SUCCESS in 50m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c3494b342594d91931017900f2bde98 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":5},{"id":"1733504c62b2f09914a6b6f4ede2b90f49cf18a2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-31 15:53:33.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4be23b8c6f09807f3364b0a80b0bf35047aa1502","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 16:13:15.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1e2e423abfde42a09c9f8884946e67ed : FAILURE in 14m 10s","accounts_in_message":[],"_revision_number":6},{"id":"64df12c3ce6d1c8bd0a1b0ca9aabf13e13a6e8ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 22:47:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ec3ef37e7744762b17a401ba10eedc5 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41be2e7907c846869f936c4be40d15a1 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/866738cae1c64d6090085d3d7e519f2e : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69a5759ab0514d53867900710a20770e : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/557a9c7fdc8e4f2a8a92b8e514e09571 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f9ce9c03b214ab0811d6ed51f714611 : SUCCESS in 3m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/46c7ab0bd9314876a974613ad4fd3305 : FAILURE in 9m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9c8c83684bc74bea9c6d5a0db326e4cd : FAILURE in 10m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aa19c8e09c6f4c2c9b5657e526e17403 : SUCCESS in 9m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/236f0755e3824912853f634f9136412a : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1ebc8b0d09714463ae453f5ed3af7774 : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24f97d8a36584c1dade3cfb27ac89012 : SUCCESS in 6m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a92b446dd24438980cf048d81a9d2b1 : SUCCESS in 6m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7c46bd88455d49f08a238a777fd2bde0 : SUCCESS in 18m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b0591cf693d249eabe36da9ad1290788 : SUCCESS in 25m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/05a3b8a0005840ae812650e0992e0e3b : FAILURE in 11m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5101832c0f0d4e6d8e3b864d827f1f4e : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cf0137542c4b49938e0bb4dcb9651abf : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43b284fe8d9048a4929269ba9e95e633 : SUCCESS in 6m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9fc1d775380c4af190037e7f7d8a06ba : SUCCESS in 14m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/798a405c669e46799bb35d61224e7271 : SUCCESS in 54m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78139a9bcad348b0b42c19629154bc1b : SUCCESS in 1h 23m 43s","accounts_in_message":[],"_revision_number":6},{"id":"a99bbd7b9d6395a06ec553ae1e519dae5257243c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-01 15:37:11.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"5704176e6d887e526b34d5c2a99d26b8162ee2c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 16:05:38.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/40a51a9e2dfa45ea8af8bf7ac922633f : FAILURE in 10m 08s","accounts_in_message":[],"_revision_number":6},{"id":"e73dc8413b37b8e6f0bc6359d4423fc2ee3dfe9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 17:06:59.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41f0d3bbf4b4c9892f64cc5182de4cc : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/609bcf4e6e934609acffb09ce421d050 : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc524da89d5d45569f2960d09a75cbcb : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9289399531814aea870eb84675a975ca : SUCCESS in 3m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/73048175d9224750a2a212afea50b887 : SUCCESS in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c8f50697b9b047f0b79eedd25620f9e7 : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cdb879343240499faf99f864037482ae : FAILURE in 7m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d8db5786a2874196b1075fcf876ba913 : FAILURE in 6m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/08a9fb82c63746c69269438cd63a620c : SUCCESS in 8m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/99d0eb9fe70141b996780238298d804c : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4c5427a12e614aa28c9740a77727e1f7 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b34e15f8cbe6491da5f551a5d76a8991 : SUCCESS in 6m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e224504044614ee1b5033e7f2322d826 : SUCCESS in 9m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/da5a17443fa74f3996b667b1ab483bf2 : SUCCESS in 18m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/55eae5e515894c5a9dddd8336ac3f0fb : SUCCESS in 27m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/17671f287ede477589e61f68332c03e4 : FAILURE in 9m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98e06012dd374917beffd08a6e23c936 : SUCCESS in 13m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fc8d81bc485a467f842a22e21d03380a : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/914a51b412dc41dc948bcc7728307c77 : SUCCESS in 8m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4ef0aa6b1def41908be73c3c2f7ac812 : SUCCESS in 9m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19fd5bb6381140cabc79c3ebc3388212 : SUCCESS in 46m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5bfc5b40987e4c548474612ea4dad1dd : SUCCESS in 1h 20m 45s","accounts_in_message":[],"_revision_number":6},{"id":"b4439b45955481fd67a55df3403a66911b6fa84a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-01 22:12:38.000000000","message":"Patch Set 6:\n\nthis was superceded by todays run, you want to abandon Hervé?","accounts_in_message":[],"_revision_number":6},{"id":"0f427423c243be14fdf14cf455fa1a9b78842740","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-05 15:36:58.000000000","message":"Abandoned\n\nabandoning since it\u0027s been superceded","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"61384c91c36a240978ffb06ec77ead11dfdca091","revisions":{"2337aed3e74b51687f42148923b491fe0a757d25":{"kind":"REWORK","_number":1,"created":"2020-07-28 06:30:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/743463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/743463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/743463/1"}}},"commit":{"parents":[{"commit":"5cd905477a7af5d691611ae7c908ba88a152fafe","subject":"Merge \"Bump pyroute2 upper version to 0.5.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cd905477a7af5d691611ae7c908ba88a152fafe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-28 06:30:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-28 06:30:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1ad04d2a9e2aa35cf9d90e00d4137c34a62fb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2337aed3e74b51687f42148923b491fe0a757d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2337aed3e74b51687f42148923b491fe0a757d25"}]},"branch":"refs/heads/master"},"a4b2aed80137dd63eff32651799f7698a9d3622e":{"kind":"REWORK","_number":2,"created":"2020-07-29 06:23:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/743463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/743463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/743463/2"}}},"commit":{"parents":[{"commit":"1454429af1df15fee960f3fdff8c0fc107f0d35c","subject":"Merge \"update constraint for debtcollector to new release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1454429af1df15fee960f3fdff8c0fc107f0d35c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-29 06:23:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-29 06:23:20.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1ad04d2a9e2aa35cf9d90e00d4137c34a62fb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4b2aed80137dd63eff32651799f7698a9d3622e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a4b2aed80137dd63eff32651799f7698a9d3622e"}]},"branch":"refs/heads/master"},"398a926d8e52a4d9b1377566a1a1f2604b2ef170":{"kind":"REWORK","_number":3,"created":"2020-07-30 06:25:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/743463/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/743463/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/743463/3"}}},"commit":{"parents":[{"commit":"31543738647d17abcb7811fd11c4b09f1fc74fab","subject":"Merge \"Drop crc16 from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31543738647d17abcb7811fd11c4b09f1fc74fab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-30 06:25:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-30 06:25:18.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1ad04d2a9e2aa35cf9d90e00d4137c34a62fb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/398a926d8e52a4d9b1377566a1a1f2604b2ef170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/398a926d8e52a4d9b1377566a1a1f2604b2ef170"}]},"branch":"refs/heads/master"},"af52a93dbd34886ceebbbc6ab28b970619238c21":{"kind":"REWORK","_number":4,"created":"2020-07-31 06:32:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/743463/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/743463/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/743463/4"}}},"commit":{"parents":[{"commit":"31543738647d17abcb7811fd11c4b09f1fc74fab","subject":"Merge \"Drop crc16 from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31543738647d17abcb7811fd11c4b09f1fc74fab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-31 06:32:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-31 06:32:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1ad04d2a9e2aa35cf9d90e00d4137c34a62fb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af52a93dbd34886ceebbbc6ab28b970619238c21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af52a93dbd34886ceebbbc6ab28b970619238c21"}]},"branch":"refs/heads/master"},"3fe7380fbc1393cb6f2c070cb671ab2076092c48":{"kind":"REWORK","_number":5,"created":"2020-07-31 09:29:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/63/743463/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/743463/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/743463/5"}}},"commit":{"parents":[{"commit":"31543738647d17abcb7811fd11c4b09f1fc74fab","subject":"Merge \"Drop crc16 from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31543738647d17abcb7811fd11c4b09f1fc74fab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-31 06:32:10.000000000","tz":0},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-07-31 09:29:28.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1ad04d2a9e2aa35cf9d90e00d4137c34a62fb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fe7380fbc1393cb6f2c070cb671ab2076092c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fe7380fbc1393cb6f2c070cb671ab2076092c48"}]},"branch":"refs/heads/master"},"61384c91c36a240978ffb06ec77ead11dfdca091":{"kind":"REWORK","_number":6,"created":"2020-07-31 15:53:33.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/63/743463/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/743463/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/743463/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/743463/6"}}},"commit":{"parents":[{"commit":"31543738647d17abcb7811fd11c4b09f1fc74fab","subject":"Merge \"Drop crc16 from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31543738647d17abcb7811fd11c4b09f1fc74fab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-31 06:32:10.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-07-31 15:53:01.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\npymysql held back for keystone issues.\n\nChange-Id: I1ad04d2a9e2aa35cf9d90e00d4137c34a62fb6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61384c91c36a240978ffb06ec77ead11dfdca091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61384c91c36a240978ffb06ec77ead11dfdca091"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
