)]}'
{"id":"openstack%2Frequirements~743559","triplet_id":"openstack%2Frequirements~master~Ic458ceaf5f5959c64dd5fca43a5eabe18367ac74","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic458ceaf5f5959c64dd5fca43a5eabe18367ac74","subject":"update constraint for automaton to new release 2.2.0","status":"MERGED","created":"2020-07-28 14:49:51.000000000","updated":"2020-07-28 22:21:52.000000000","submitted":"2020-07-28 22:20:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743559-1595974811355-be75a880","meta_rev_id":"5091e8b1051b04eecbd584eef1d9d0dff7d749ad","_number":743559,"virtual_id_number":743559,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 22:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-28 14:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-28 15:10:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-28 15:10:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 14:57:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-28 15:10:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-28 22:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea00c7d226c9dec97f480f908ae2fc9ccdcf73f1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-28 14:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b391bbda4be8395ae45bbedb840b631becc9c38","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-28 14:57:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc3030fe8de5656ba4e278468d924e068abb5e8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 15:08:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"ae71d5f272e9489ea5b2d0b1964e770c4fa1130c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-28 15:10:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"423fb55a20a748df726ed102e4ab03587afed845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 19:01:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb78cf35a1d74b55baf113dbf31fd2af : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92c9e44290e41feab9b85aa19e32417 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9379043163e4350a181e9f0eb134add : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db4ca991aa8e491399fe8d5d3462a56c : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09a3d44094e94b579b2a768d21ea742f : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cdc51b631cec4d00aac51672c6e2d72d : SUCCESS in 5m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4b67fa9400884dacb92cb85bce2e1a03 : SUCCESS in 8m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0d207026474e4241870843ad118b27dd : SUCCESS in 8m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2c49fb2771e8452e8277e3ec5245046d : SUCCESS in 10m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c53c63270f3845cd90e37e00f38c6f53 : SUCCESS in 5m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6ef1003ad1c1437e8605a4ea2cc2b7df : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c55be2d1ec774294801137fc0a895da3 : SUCCESS in 7m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0ade632083f34374856f5d745ce80531 : SUCCESS in 11m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b7f200ebff254b8bb5e233ffba9e5324 : SUCCESS in 13m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/070924f2b0b34f5c884ed6742e2965b3 : SUCCESS in 26m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7429af52390844a6a332e9ac31ebc349 : SUCCESS in 14m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6879b701cf814188be6530adef620155 : SUCCESS in 25m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a22c389ba5f74551a8a8c102071c065e : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2edc0189f464a7faa667d7e241ba36e : SUCCESS in 6m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aaf48bbe8bb64b33a512eb7cbe82a60f : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/915955df832045d9a602cc03b9cbf3b9 : SUCCESS in 1h 10m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/715526efa313460a835dc9343a330e2c : SUCCESS in 1h 13m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3b386a2598e3303f6c6e494606793b558544b0c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 19:01:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7489cd1b43744a65e697b98ef23de954bd69cae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:20:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07c2398a912471783ce60bd640f1cef : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af9946001034c2b83c8118713cb0049 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3fbdb9a663841789fe263b085e60b65 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cccde206b4fb4d12afd80eefd51a54c6 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2d47d4564e04a50b9dadb8f87113de8 : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94e98db759514fa2b2e5cd2d9672e848 : SUCCESS in 7m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9fd90a0dc4e94bf29cbd955f1ce0d744 : SUCCESS in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ef7a1845ce6435e9cf01637bf32e2c5 : SUCCESS in 12m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7adee3ad714640009bbcab61516be21b : SUCCESS in 10m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/56a4ef10e36a449ea3c109b74e8d6a9e : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7223fd911877488c9e498d78325087ae : SUCCESS in 13m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ca2bb8b3a0d4a7f88fc176a9d3e28d9 : SUCCESS in 8m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/433d62352dd9489bbb4ec7384075eeb8 : SUCCESS in 11m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d947dfe552be4ceabb1eec91d3e3a232 : SUCCESS in 25m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/884fc610045c4ed9b5a6998a6a9a9ba3 : SUCCESS in 31m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/160b2378493345b0b3761c76ce94bfe3 : SUCCESS in 13m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef9ac200b0e34e18b75382d5f2d21c1a : SUCCESS in 17m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8aaff320f2f84bf0826b34c9e2518072 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5890198112774745905979a235d425aa : SUCCESS in 8m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/67e15a5fc07c4a868e494ebad9bfd2d1 : SUCCESS in 12m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ea69588fee642a79d70e42c808b46c7 : SUCCESS in 16m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffc18b2a9b75464e871995f1d25140bc : SUCCESS in 57m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9720543535b45baaae4272c914b3751 : SUCCESS in 1h 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cbab54e7cb0006bd92217ab77efd5a95c222e11a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:20:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"edafe80c9c7d291f2f6caba78e38844b55ae2e34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:21:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a024aca7ce240d882eea9be97ba52f4 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"902a13f52971a119112878369c96492ff65d7a36","revisions":{"902a13f52971a119112878369c96492ff65d7a36":{"kind":"REWORK","_number":1,"created":"2020-07-28 14:49:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/743559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/743559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/743559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/743559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/743559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/743559/1"}}},"commit":{"parents":[{"commit":"5cd905477a7af5d691611ae7c908ba88a152fafe","subject":"Merge \"Bump pyroute2 upper version to 0.5.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cd905477a7af5d691611ae7c908ba88a152fafe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-28 14:49:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-28 14:49:47.000000000","tz":0},"subject":"update constraint for automaton to new release 2.2.0","message":"update constraint for automaton to new release 2.2.0\n\nChange-Id: Ic458ceaf5f5959c64dd5fca43a5eabe18367ac74\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: If196e5a7a236215a626c15c70ad0b422dcbfa1ed\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/902a13f52971a119112878369c96492ff65d7a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/902a13f52971a119112878369c96492ff65d7a36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
