)]}'
{"id":"openstack%2Frequirements~743560","triplet_id":"openstack%2Frequirements~master~I7b4b873944bac3b00dda9ec75ae26427866d8288","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7b4b873944bac3b00dda9ec75ae26427866d8288","subject":"update constraint for debtcollector to new release 2.2.0","status":"MERGED","created":"2020-07-28 14:49:52.000000000","updated":"2020-07-28 22:37:37.000000000","submitted":"2020-07-28 22:36:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743560-1595975767471-f4f80a02","meta_rev_id":"c8e75bd897e28e1cc6ba7e85b746f71fe7b7ace6","_number":743560,"virtual_id_number":743560,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 22:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-28 14:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-28 15:10:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-28 15:10:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 14:57:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-28 15:10:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-28 22:36:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"846e70ab39977144f2035c0381a77b5e40a171d7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-28 14:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc474e97fe5391c54dd281ffc94e3809ec2820ca","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-28 14:57:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3c041d3ff04e30169784fa4ad0c99cc208a2ae5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 15:08:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"1cf7be475004f22080f8fc6f11fba208ec283d14","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-28 15:10:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52c961bd3a2bfd5f206d97c63666a505683b9a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 19:28:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440a3e84b9944cb4bd070b6b06480640 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8781cbf6ec90457b8a323b763542f6d6 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbdddc8605ab437a88ce59c7a13039df : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bd93dddce38492d842b0cc5f01ce170 : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0311986e7c64defae1e639655769aae : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1b9be15cd324260a17cbe90421bb526 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d646ff3a656145d5b58bc8ca16b3c85e : SUCCESS in 6m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3c159bfd5d47443489ca254f44ce8748 : SUCCESS in 11m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c06df1c619544d7e92cb0dbeb5813af2 : SUCCESS in 9m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/48d21ca986114af39cc4743f0cd8a066 : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f0d9403d7a15478cadad71d235ec1b56 : SUCCESS in 17m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/620405f383e8436bbd0d27387fcb8fa2 : SUCCESS in 8m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ffdd79a0130d43a1b9f55ac87a0f90c0 : SUCCESS in 7m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3b35f385d99543ccaac68eac43eea7be : SUCCESS in 20m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4d17f05d6d1c4d76ab12470b0682f041 : SUCCESS in 25m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b300dea081d94589b3cdebc1b097971e : SUCCESS in 15m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb94037222b24bbbab43e34d480bea1b : SUCCESS in 16m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c8315360a254411e8cb678bda607198d : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f2f1bbffbc124a0cbb1a7e062e7725c6 : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/181c597eefbd4c839991cf9c47ba70bc : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/369c95097d62479c993afda4136414cc : SUCCESS in 54m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0be747420ef4348813dff669d83e9df : SUCCESS in 1h 35m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e7ba8f3ef2384beb0a2ef07ac0ed0f30250af5b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 19:28:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a5ccc732752c50221ec252a860c60da4f31fc1c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:36:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9778cd0a57854673b0cc1a6af83ed073 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72fa0dd7fcaf4885913cd63868d72cb4 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d77c4517b13640e9a1f3d2e0f0b7807d : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7955b2c5cffe43ff8f29c07c8489c217 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66078ecd90c54b189164e64032619a3b : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/19479a655e994dc7b7598b769cb8f148 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0e021cd24de94e7bb35ad4befa484f40 : SUCCESS in 8m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/279937535a314950aadcc2f70c81e08c : SUCCESS in 11m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/96b519f47073469da0df13f201997f56 : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dbb2d8254568441395f7e8651a137d56 : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/67e973ba2bd4448b8cb7115290de39e8 : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8eac4c85e2f3469d92e83ec8c7589020 : SUCCESS in 9m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/75561f75dfd449769dcd290781846057 : SUCCESS in 10m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2fc97ac0f6cd463ebe377b8c94b09b87 : SUCCESS in 20m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/82a7ea0e9996483f945e85ff04cde719 : SUCCESS in 27m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/65782b0fcd014bdd85d563e665d76d3e : SUCCESS in 14m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d85eca6e1211468bb3d5061267a9aad6 : SUCCESS in 19m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/11f7241341b34b0ba7d21d1357876e05 : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/096e78a109494354be3c169368fe63a6 : SUCCESS in 6m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0148d871c0554d4896d3b6e9a8ac38a5 : SUCCESS in 12m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bfcbc3f3471345e4bc710d66b397121e : SUCCESS in 19m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c717b9bfa2c4f538d80c19dd84525aa : SUCCESS in 53m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f922c9b4f9d47869f22a65184707e52 : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2f30c65eb1151468a26e052b841dcd2b21dce6fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:36:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"de50ac451cbedcb2e5923b2a2bfe57cdf5a8e656","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 22:37:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb25a2ebc3e4da9a21512c033c1c5d2 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44f264bbb8bc8a40ec4b673cb56c6504ba737748","revisions":{"44f264bbb8bc8a40ec4b673cb56c6504ba737748":{"kind":"REWORK","_number":1,"created":"2020-07-28 14:49:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/743560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/743560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/743560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/743560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/743560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/743560/1"}}},"commit":{"parents":[{"commit":"5cd905477a7af5d691611ae7c908ba88a152fafe","subject":"Merge \"Bump pyroute2 upper version to 0.5.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cd905477a7af5d691611ae7c908ba88a152fafe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-28 14:49:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-28 14:49:47.000000000","tz":0},"subject":"update constraint for debtcollector to new release 2.2.0","message":"update constraint for debtcollector to new release 2.2.0\n\nChange-Id: I7b4b873944bac3b00dda9ec75ae26427866d8288\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ib9a30a954f01279df3bd4522d4324cbf7e361839\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f264bbb8bc8a40ec4b673cb56c6504ba737748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f264bbb8bc8a40ec4b673cb56c6504ba737748"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
