)]}'
{"id":"openstack%2Frequirements~743841","triplet_id":"openstack%2Frequirements~master~Iebf88d956582dc88fd3e1c9acc78f455297533cf","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iebf88d956582dc88fd3e1c9acc78f455297533cf","subject":"update constraint for os-brick to new release 3.2.0","status":"MERGED","created":"2020-07-29 20:22:15.000000000","updated":"2020-07-31 13:16:00.000000000","submitted":"2020-07-31 13:14:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743841-1596201278500-a39be2e8","meta_rev_id":"ca3dbf885614d50528689fc6aad20acd38ffcfa4","_number":743841,"virtual_id_number":743841,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 13:14:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-29 20:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-31 09:26:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-31 09:26:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 20:22:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-31 09:26:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-31 13:14:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b25003f1f53dfc0b7652e14971a523b47cf1517","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-29 20:22:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88370fa91dacf5af14a2e0de52edbe870d47ad24","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-29 20:22:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7fb7f01055415ca0e35daa488855b4c3f577f8ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 20:25:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"d2f1de6313dc4749ed22a468c65a37d5aef73fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 01:16:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab24f6bf91724846b785e1fcd69f6568 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/192ec8b736c5476999952a39ae6f726a : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39ee715f2f0d42c1b41d38c6fa0b8df1 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46b3b42e87334a9cbba08649b84fcd10 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04a765b5f50346fe9cf67952a5bb29e7 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ac22826105a46d99ff8bfbc60169ef0 : SUCCESS in 3m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ebde95fd7e1046ba83d9f4414361d11f : SUCCESS in 21m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ba5dd4c739c4204804f79f571fafc0d : SUCCESS in 10m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/614a99538b464e0395b040b728d6c988 : SUCCESS in 9m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c0a2d9861c054a41bb2d6131df49143b : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/417317d5e7d44cd0ac4782482aef7adc : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5f0e425f22944cea26cb5ff74a4f66e : SUCCESS in 6m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/12ce98c65c2d45daacd26c47d8b180d8 : SUCCESS in 7m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2654352fb1804d38b3703ef1a4e68854 : SUCCESS in 12m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/51d1aa005f9c4ef78575d80b436da534 : SUCCESS in 24m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/21a0630c5b7545a38508fcb94292b3b0 : SUCCESS in 12m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f79c75f20817461b9155243f4dfb48fc : SUCCESS in 17m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3e2a89b291f04fe2b0e90e267665607c : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c90c3a92e2f49a0b94959c779fdccb3 : SUCCESS in 8m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0d88b9dcd37c46daa3ae553077db70e5 : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/171b392a80ca459582024fe03c2e7a78 : SUCCESS in 50m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f86505c3e9d14ffc94174200f97a6c6d : FAILURE in 1h 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6aebf09c4ae5a61b52a7735dafd1e2bd794e237f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-31 09:26:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82a97444f0f4c9c9d62b25504514a9b1584f1789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 11:55:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f64276c4d4b472da4afc40f9e53bd6e : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f13661675804f3781b328e9255e52cd : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c2fa9d5d06049b9bb611137880e5e6a : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a61d94532b4f483e84c874796d101ad8 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4aa780237634ff7af81dc3d73481091 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4541bbcc6bd4001b85f25b75d5c8549 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bb9e5ae1029b4b93a3bc0823779f05d9 : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb81b8a822d4466bb0fd4547eb187fc1 : SUCCESS in 10m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6f45c99a85884a0d8e5a481c113ccdc1 : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/34d854df99df4627811987698af2295b : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8bfdb598dafb44a98e5087d6ec2eed2a : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c24ee3ca802c4f9f945288b2ad3f6430 : SUCCESS in 5m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a0c4db6e414a4aa4a2837915eb6468bf : SUCCESS in 10m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e7fab5424a3f4a199e31d1a4c5876cf7 : SUCCESS in 18m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/28c8192835384238b1718e6160768c10 : SUCCESS in 24m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6625957e17314654b85a1982783a6906 : SUCCESS in 14m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c012fdee97949baaab9584e1675f6f1 : SUCCESS in 16m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cfa2dcd9551545358a27f4a3bbe67107 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40b58d8805f04bb1a1caca26983ca38a : SUCCESS in 7m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2d19d2b402ec4b7086ce0b837721b80e : SUCCESS in 8m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa61a32a7c634298b7a6e0d1b8c1dbe5 : SUCCESS in 50m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f78a2909bc148a8bdf3aeb12ec94787 : SUCCESS in 1h 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b3e0ee4293b1d3e63384d0029d789f28c8709596","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 11:55:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83253ae598d0ad9abdf3ce88e0d7b61afa13e270","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:14:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8e4cd9ec2a5b2a4308e40280584e71ca60a5c006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:14:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81bc65993ce4bf0b98c3ba5142ee48d : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bceb1cf651a4418da1b57c1a0e33ec42 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55772e77a41a4588883a4b1d34701534 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e0d276d25e34a7caa54c88d881f9180 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/541a4ac8b9934cf8b332967dbb47a73e : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9784d564a24642038045d70d135c6662 : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2b5419a6e1245dd9994957823158717 : SUCCESS in 8m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fd8f76c529ee422b9a9ccbf5be816e3d : SUCCESS in 9m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/952cfee707044cf28703e8762ba3e79b : SUCCESS in 10m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/feb1215b01994116b65f812ac5c42dcb : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fcd80c4925ca45fd9d88269195d914b7 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5583c23414c448e6b4a364ef403dd0a1 : SUCCESS in 7m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9565f9f1c10e4a1faf44c7ea7066e553 : SUCCESS in 8m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eaf7cef5e9454e9293f3c73444a60db0 : SUCCESS in 18m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/19e26aa975cd442fa63bb5a15131e915 : SUCCESS in 26m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f76c7af47e8c4ae3bedc292a6d18ca30 : SUCCESS in 13m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be085704f72a4b38809140c4234924f5 : SUCCESS in 16m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/75cf6509d52448ea8e327ba0660ff354 : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29bde0a1c30e4fc08f7ac938d2cde8a6 : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a73f37c8493b4ed687802bda63e7a8e5 : SUCCESS in 12m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/385364c698ff4cebaef4413ce1438c05 : SUCCESS in 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3db4740d25e44ca69fbb1e7e259ad75f : SUCCESS in 50m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec55ffaf5a964e98bc91b2cb5c1a0f93 : SUCCESS in 1h 15m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ab0285956a9e66a9f98dfb66b72a080ebbb695ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 13:16:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d8b9ea02e9e42ec8140946795c20479 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7959d0c218949981b8166f9cb5b168f681a7c27b","revisions":{"7959d0c218949981b8166f9cb5b168f681a7c27b":{"kind":"REWORK","_number":1,"created":"2020-07-29 20:22:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/743841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/743841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/743841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/743841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/743841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/743841/1"}}},"commit":{"parents":[{"commit":"0e87b2558febdcd49ee8319445d3251a4da9a158","subject":"Merge \"Remove hiredis from global requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0e87b2558febdcd49ee8319445d3251a4da9a158"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-29 20:22:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-29 20:22:12.000000000","tz":0},"subject":"update constraint for os-brick to new release 3.2.0","message":"update constraint for os-brick to new release 3.2.0\n\nChange-Id: Iebf88d956582dc88fd3e1c9acc78f455297533cf\nmeta:version: 3.2.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ib10d7f66d7e83d3556f3288fa0b7e623125b4855\nmeta:release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7959d0c218949981b8166f9cb5b168f681a7c27b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7959d0c218949981b8166f9cb5b168f681a7c27b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
