)]}'
{"id":"openstack%2Frequirements~744176","triplet_id":"openstack%2Frequirements~master~I2cb38f99533ae17837a57381c6e5f428f1fd6963","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I2cb38f99533ae17837a57381c6e5f428f1fd6963","subject":"Updating pecan to new release version","status":"ABANDONED","created":"2020-07-31 13:58:56.000000000","updated":"2020-07-31 15:41:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4101fb9499499d9abefdccda06661f38d42953bd","_number":744176,"virtual_id_number":744176,"owner":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check-arm64","value":0,"date":"2020-07-31 14:14:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2020-07-31 14:05:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-31 14:05:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 14:05:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-31 14:05:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-31 14:14:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d13b6399e97a1bbf7a2832a1532d6f923a8cfb0d","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-07-31 13:58:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba67a3b6c9a31737a031149c926e60b20f5526aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 14:14:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ba12c1348abf4faaaae97853012b1c60 : FAILURE in 11m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4101fb9499499d9abefdccda06661f38d42953bd","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-07-31 15:41:51.000000000","message":"Abandoned\n\nupdate already included in bot general update:\nhttps://review.opendev.org/#/c/743463/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ab927da108648b42f3ce9d77e534cb2c9285cd7f","revisions":{"ab927da108648b42f3ce9d77e534cb2c9285cd7f":{"kind":"REWORK","_number":1,"created":"2020-07-31 13:58:56.000000000","uploader":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"ref":"refs/changes/76/744176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/744176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/744176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/744176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/744176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/744176/1"}}},"commit":{"parents":[{"commit":"885a1cf4e3208a3e622d513a1d55996e23958728","subject":"Merge \"update constraint for os-brick to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/885a1cf4e3208a3e622d513a1d55996e23958728"}]}],"author":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-07-31 13:58:17.000000000","tz":-240},"committer":{"name":"Dan Radez","email":"dradez@redhat.com","date":"2020-07-31 13:58:17.000000000","tz":-240},"subject":"Updating pecan to new release version","message":"Updating pecan to new release version\n\nThere is an Accept header handling fix in 1.4 that we\u0027d like to\nconsume in neutron.\n\nChange-Id: I2cb38f99533ae17837a57381c6e5f428f1fd6963\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab927da108648b42f3ce9d77e534cb2c9285cd7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ab927da108648b42f3ce9d77e534cb2c9285cd7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
