)]}'
{"id":"openstack%2Frequirements~744247","triplet_id":"openstack%2Frequirements~master~I269d399359868142a968dd981db75b2ecf4c920b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I269d399359868142a968dd981db75b2ecf4c920b","subject":"update constraint for oslo.privsep to new release 2.3.0","status":"MERGED","created":"2020-07-31 20:15:19.000000000","updated":"2020-08-01 01:05:03.000000000","submitted":"2020-08-01 01:04:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744247-1596243841461-5080a3ed","meta_rev_id":"26907cabce5706b3dbc1b05b5cc28adaa92d4535","_number":744247,"virtual_id_number":744247,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-01 01:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-31 20:16:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-07-31 23:25:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-07-31 23:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-07-31 23:25:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 20:16:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-31 23:25:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-07-31 23:58:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-01 01:04:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4957606395f4a7de13c595a17b8ade514387548b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-07-31 20:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59ce913073b95fe543de0a40751f4e4801016e6b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-31 20:16:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab16b18bbd8e4fa4e03db5e1a3a69719bc9bfc51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 20:32:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fd3c02bb44cf440491e264fd6b0624b2 : FAILURE in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5df112f9b93ba30738a6d38b8e4683634f35c060","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-07-31 23:25:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"beae7771a5487ee0b5abc40eb2902d59d1e4b411","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-31 23:58:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a5fb28c66aa34af8a2788cf4548b04ee8863b4f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 00:06:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e796123daa45f895d27f9400db412a : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85537ee01834af1b22cfe06a24ec52a : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/498ad2e626a44ef08141545491ba9269 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2dab59a89fa43d29b24657b54709a85 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef718ef79eff4321a36a6821779827f9 : SUCCESS in 7m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/38b316bb3bb94c9fa59a6a12c98ac5d5 : SUCCESS in 3m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6bf74889ece4b17b80ec1d536ee263b : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d5270e3597294daa8e49a4862f526886 : SUCCESS in 11m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6653a729eacd403f913bc7eb8c42b1b1 : SUCCESS in 11m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ce8912e14f794ebfa024f21b0aad2d15 : SUCCESS in 6m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ad6710c13f6040d585ba2280948dea12 : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b765ca05e0764763b5c84d39ec9d750a : SUCCESS in 8m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1d5dda394cd94ecc8978c9f4da28a4a8 : SUCCESS in 6m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ab9a9788955545248a89a304fbcb50ee : SUCCESS in 12m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d1976bb440fd45aaa825631aac2f514f : SUCCESS in 30m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/853ede4bd89c4f17a6e92bfcd84aab69 : SUCCESS in 13m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9cd6bd40c1d14f03a6b798a02858db18 : SUCCESS in 23m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/acd163c2a22f49b483c9737163d57b7b : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61d412582e154233a4c2998f00750044 : SUCCESS in 6m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/16b943abfd3742428bed7c575dcfb81c : SUCCESS in 13m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd46e5764d5d41f48248759ba0bf47c2 : SUCCESS in 48m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c54f958ddd44f658031a7db895dd832 : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":1},{"id":"41965d6ee9023498d56d9f2b01895ccdee6b248e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 00:06:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f0b43cc97c245c02c818251fb418148f415265f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 01:04:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"19ea3dd50b56f3e25a49ef902160176481808536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 01:04:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a909b0f28bee4bd2b11b44b24626da6f : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f1175f0c5eb44ffb1bdb0b410ddba81 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aac4707f19b0471bbbce602e06c84c30 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6045f2a5343845378abc5cd1a1dd5130 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c1d37a53d2a4f9bb073c1ae306e1c0c : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb4c6f29cb834bd085e94cfbf9ec5cc0 : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a2b77e7363e471baf587b09dd3545a0 : SUCCESS in 7m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b2dc2a14ba9a4ce8aede521028f8dccc : SUCCESS in 11m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/10f760f015024c49a6e129118c7cbbb0 : SUCCESS in 11m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b5606d4453cc4d888c2bf63de9d72b6a : SUCCESS in 4m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/19fc20af16cb468da342e480b43e9f2f : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eeabdf29f2574af09aac79c191900d39 : SUCCESS in 6m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/63a2da478919486996fdd7747e4ea4a1 : SUCCESS in 7m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f11147662f674181abe93bfd383b662a : SUCCESS in 11m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2fe38af48004475fb5a4bc4ae623ec53 : SUCCESS in 24m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/94fce5591def4e77a1ff4935b1738f99 : SUCCESS in 13m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c60b6441b7fb4b439e613fdaa7d46ced : SUCCESS in 15m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/51a36f62b9e54a84b672c17ee9fee9ce : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6cd7be25e1a44eb8780d1c0b1c96ebe : SUCCESS in 6m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0dcca4c3ffa54315b165a91728c4ede3 : SUCCESS in 13m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/133b2f7b9c384650baf1798f591e7c4a : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f0c5a0b8eab4db2900e1d094810a26b : SUCCESS in 48m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45b715b55bcb4a5d8dfd3566e6f6c4eb : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b9e78661fcf865627ee8e5fb60df333e574469ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 01:05:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/514f3f8d4d604189bd521f981b8807ef : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d58eb44fd356cbe3022fc8cc3cc02fd06da5fc23","revisions":{"d58eb44fd356cbe3022fc8cc3cc02fd06da5fc23":{"kind":"REWORK","_number":1,"created":"2020-07-31 20:15:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/47/744247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/744247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/744247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/744247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/744247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/744247/1"}}},"commit":{"parents":[{"commit":"885a1cf4e3208a3e622d513a1d55996e23958728","subject":"Merge \"update constraint for os-brick to new release 3.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/885a1cf4e3208a3e622d513a1d55996e23958728"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-31 20:15:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-07-31 20:15:17.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 2.3.0","message":"update constraint for oslo.privsep to new release 2.3.0\n\nChange-Id: I269d399359868142a968dd981db75b2ecf4c920b\nmeta:version: 2.3.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I673491f8ee38ab4b2ffed1ef8629002b4fd8a896\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+1: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d58eb44fd356cbe3022fc8cc3cc02fd06da5fc23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d58eb44fd356cbe3022fc8cc3cc02fd06da5fc23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
