)]}'
{"id":"openstack%2Frequirements~744274","triplet_id":"openstack%2Frequirements~master~I05d6901be5f726f68f52a4963a733aea11cdb77f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I05d6901be5f726f68f52a4963a733aea11cdb77f","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-08-01 06:30:49.000000000","updated":"2020-08-01 20:46:40.000000000","submitted":"2020-08-01 20:45:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744274-1596314705644-c0251aed","meta_rev_id":"178d1888d77860a49701bacd4284a9dd0a0cb17b","_number":744274,"virtual_id_number":744274,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-01 20:45:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-01 18:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-08-01 18:38:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-01 18:38:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-01 20:45:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16d72d6b6b2d8ad6455c66ac9ab3baac30fef20c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-01 06:30:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"573f473ab09c6977d5af03db1846e13d4c2a3634","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 06:55:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/95de2a6315774e96a60d08ce1ad4d7f1 : FAILURE in 20m 11s","accounts_in_message":[],"_revision_number":1},{"id":"5a01c8fd29b04a89014f5fa3eb396289d22fe34d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 08:28:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4bd26b16f014b41b72ccb93722579cf : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3e72a58b1e48cab7c82d2132ca29d5 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35c069043b664ed7ba987e17301b2efb : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f4e5018e24943b29cdb1cb23b1467e0 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d17023315e54ab88323a64b5b81d6b1 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a95f5252dbdf4ec8b8f61ee0fd422fb5 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a646969675b14a0d889d2dea7d2c240b : FAILURE in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b95caec6859949d5ad499c8cb480cf06 : FAILURE in 6m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/94ae4f901f544907ad2b38df2485d042 : SUCCESS in 8m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a45c084eb2464ec49d0a82233670f791 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/71836d1e442a4738b83ff3d4a3d83b21 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c981537e15b44b49ff8d04aecae670a : SUCCESS in 8m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4dd4b1dc788746a5bd1e8f4c55059b90 : SUCCESS in 7m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/df9dfa782dc44ff59b60fa0b666e5b4c : FAILURE in 10m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/48c48f4113ab4c2cbc601a2239a5e724 : SUCCESS in 28m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/93dbc666be5540258765847f3cd7ecb2 : FAILURE in 12m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57dd0c41af4943568488a407917523ab : SUCCESS in 11m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/35abc55b6fb14520bb12986c76f9223f : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1c761bf6bfa745d7b026a14b8703a1ba : SUCCESS in 6m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/02a2d000d97d40ea8f0dca7db1f5587b : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/233126bdbce94677a7024f46629105d5 : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36ce194071274666b7452c57798b5eda : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":1},{"id":"46f32f0bbf111b7980f0d7d59abe0e8d45e4a990","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-01 17:59:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"165b109f8da9a994825f8d14792186cdc4b21bc9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-01 17:59:03.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"a558d7df72232bc43603aad501b693ea8531b819","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 18:18:46.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/20df71307bb349d49fe4001f29598f32 : FAILURE in 15m 44s","accounts_in_message":[],"_revision_number":2},{"id":"432422a897abab1c883395a4e60029554cf8625f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-01 18:38:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b9d431828468dff4db3b625b4518cc9de8b2601c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 19:27:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155cc62dfb5e453d9ec52071748b7b19 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0551d2f48cda4cc5b111460b4305407a : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21cdeb07eeac4e929a132647be7372de : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cdceeeff9a54fa0b89ee430fee31537 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ed71b18faec44e58cc84d90315332c7 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5c619b33b8624fc990150e714514a31b : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9161d4d57c8486ba93e953edfd39aed : SUCCESS in 8m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/941fa8c1192b42c7a648eaf86b032161 : SUCCESS in 14m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/90d869e2c0c44902932f10a42d470c58 : SUCCESS in 8m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/58186f05b0d6496991f6af081976616b : SUCCESS in 5m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/65e9b3d8a8644f3a818839d826171687 : SUCCESS in 6m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af54f33550e0402ca65fbb70bdd2d420 : SUCCESS in 7m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c90778c6cd034b92a514677512fdf7d8 : SUCCESS in 8m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8666e78b7c8946c1baaf1f88350eea5c : SUCCESS in 24m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/478acf7b24854cfca302df6c2a068b6c : SUCCESS in 23m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c69d74da41cf43e987dbb30f6e22fa81 : SUCCESS in 12m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4bfe0341d313456ca8b5444682d8a602 : SUCCESS in 15m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9b3ef31eee824aaf98944ca4e1d3be84 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff78c35fcf00468ea690587db7fcc3be : SUCCESS in 6m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3df68a5ec3bd4db7b3fc94671463db34 : SUCCESS in 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a116a93de784f01b716f88b49536fe4 : SUCCESS in 50m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae273c5a0f15409ab0f7f92b4a70beaf : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":2},{"id":"22e8a5ca4af7bdf6e43782bfd454555be56b2eb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 19:27:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7b49cf219a8bc5b420abc0174936fea7ffe48e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 20:45:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e5a99f2efa040ab92d99818ae856bf2 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f65124e7403461b8296cb07e158fcd5 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5fd9c0ce17e4bd38303ca508c18100f : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6de7cea6131401e83bae249ff3432af : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0959f795a8d4dea96d266d739ac51a5 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/394022260de244dbb62f76adfe2da4ce : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6c4074fbcd974741b55dfee8a985d3f4 : SUCCESS in 6m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/caa062fcd4de4d788ceec215a60cb57d : SUCCESS in 10m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/83f2df232a7c4433890834acaacdb8d6 : SUCCESS in 7m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3146fbbbbaec4c79bba63effa10d34dd : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d060a86076ab4ba2858796b043ddef0e : SUCCESS in 14m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d22645e3040c4e8fa912bddc7bee9a74 : SUCCESS in 7m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6129dd6109044db7b361e179b64f2bed : SUCCESS in 6m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9756aad81d4d4a10b79140e48efc661b : SUCCESS in 21m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fd81ee5328974bb890e4d17d1ba5efa6 : SUCCESS in 26m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ff911d93f19a4c3ebee4dcab803df5c7 : SUCCESS in 8m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/60b458f2220b4776b9de7a8b1cc0b369 : SUCCESS in 15m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b9d6ac72c6b84ce9b6de3b18cc0475c4 : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b6eacb77d814f75a7d5cc89f870b3b5 : SUCCESS in 8m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4cf4d93e301b48a19003089ecfc0f13b : SUCCESS in 13m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/66579e191ddb4e949cfa76eb0777cf88 : SUCCESS in 9m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b28e4a60996f41f5b4da6c3fd6d8a72b : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d51da22a1e04757983e582874898370 : SUCCESS in 1h 13m 03s","accounts_in_message":[],"_revision_number":2},{"id":"c5d75b2da042968d6e22f8ae09d0538afa138bed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 20:45:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"178d1888d77860a49701bacd4284a9dd0a0cb17b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-01 20:46:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ef993fd17148bb8286a0c29a7847bf : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7321518cebcd366d6d827ef83c6277ee729b1b49","revisions":{"aa9cc534b9a437872c4a9d55122294f5d6511795":{"kind":"REWORK","_number":1,"created":"2020-08-01 06:30:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/744274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/744274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/744274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/744274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/744274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/744274/1"}}},"commit":{"parents":[{"commit":"d36d572f5b2ecba84451ce9a4209caede8bf2d06","subject":"Merge \"update constraint for tooz to new release 2.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d36d572f5b2ecba84451ce9a4209caede8bf2d06"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-01 06:30:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-01 06:30:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I05d6901be5f726f68f52a4963a733aea11cdb77f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa9cc534b9a437872c4a9d55122294f5d6511795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa9cc534b9a437872c4a9d55122294f5d6511795"}]},"branch":"refs/heads/master"},"7321518cebcd366d6d827ef83c6277ee729b1b49":{"kind":"REWORK","_number":2,"created":"2020-08-01 17:59:02.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/74/744274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/744274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/744274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/744274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/744274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/744274/2"}}},"commit":{"parents":[{"commit":"d36d572f5b2ecba84451ce9a4209caede8bf2d06","subject":"Merge \"update constraint for tooz to new release 2.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d36d572f5b2ecba84451ce9a4209caede8bf2d06"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-01 06:30:47.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-01 17:58:31.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I05d6901be5f726f68f52a4963a733aea11cdb77f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7321518cebcd366d6d827ef83c6277ee729b1b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7321518cebcd366d6d827ef83c6277ee729b1b49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
