)]}'
{"id":"openstack%2Frequirements~744487","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Ibb78aae930bbb98025d806ec4ca9ada04d8ce012","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"Ibb78aae930bbb98025d806ec4ca9ada04d8ce012","subject":"update constraint for oslo.messaging to new release 12.1.2","status":"MERGED","created":"2020-08-03 14:23:35.000000000","updated":"2020-08-03 21:46:34.000000000","submitted":"2020-08-03 21:46:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744487-1596491194295-bf775c37","meta_rev_id":"7a2c24e31e38d5c40012e498a65b01a476672ab4","_number":744487,"virtual_id_number":744487,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 21:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-03 14:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-03 20:15:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-03 20:15:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 14:35:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-03 20:15:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-03 21:46:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a455f7b2827cdf7c7c190acca6f99e6d10afef5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-03 14:23:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c53fb6dcbe7e29de3a906de22d3a48f039b6b73","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-03 14:35:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c1866c616db2de8bcc74ab840c0ee5dc1295337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 18:42:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b710a0d4d7486986e9c9826d9c3514 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ddb72eda6d341738f904610a99fac84 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38650488aba94e4b80fa26589f7bd182 : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ebc7c520666474b996abc6cf9cd2070 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e2b493cbae441f0bb849b3b362da7e7 : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/18a5a931f4124ed08fa7f6bee51c4348 : SUCCESS in 8m 32s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e982b20dc7554a0fab52cb534fde140a : SUCCESS in 11m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/75a5b7ae8cc14b67bcc92f7ef9c681b7 : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/165de1f3355e464290cae18a6510a8ab : SUCCESS in 7m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1697e2196ff6477eb5a8fa81cb4cbb73 : SUCCESS in 7m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7aa10080c0314f0aa778b060020ed338 : SUCCESS in 15m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9c12a838c3e640c3bfaa8a7732ba55c4 : SUCCESS in 30m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5b57dc371712458ca05f08e8cf5938c6 : SUCCESS in 15m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6a140c24666b4f05a7f5bb5c5603e374 : SUCCESS in 14m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/11e0c4b4d3c246c8be77b45c41e6d222 : SUCCESS in 18m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f6da4e96a37b42ef9dc383bede3962c8 : SUCCESS in 4m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ebf15781a63b42bcbcbed39ee3a997fb : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4700ba63c14442baf9dce82e05713e1 : SUCCESS in 6m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/567b8decc0214c57b70b3271fae50c42 : SUCCESS in 11m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d67084817e3b4267b94387c83a9d0f10 : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0122d40156145eb9f40b73d9a16e9f1 : SUCCESS in 53m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd5d142593b14f1481a2d511b6324862 : SUCCESS in 1h 20m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba15da41ca40499584f8d95e04567805 : SUCCESS in 1h 22m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/cf68fd3a54ea47308434b29fbbee1bc1 : SUCCESS in 1h 05m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1157d3c888f475665bf08eda7b780444e36e1349","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-03 20:15:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"961b81046fffccb1b7452324d5f51121f91fe2b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:15:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2df9eeb92f3b97be6e0c2f8c56fd62c9ba1faca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:46:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/471c9075fe884409a6b79f7d73bdd39e : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e9f2b5a26024c92801775693b81568c : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68dad8f80ea2430a94b41a55e499c07b : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6fb5f043299467ebf919235aa3ed2f7 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fa97494e30714f20808729e6181eff6d : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/491a4e238c30426f80cfc1b77037fe4c : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7486ad3c5dc94791b6fe759ac9b257a4 : SUCCESS in 10m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/02ffcbf370bc4d8f8449cc9e76ab6ffa : SUCCESS in 6m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2174ba6a99a144748722b40152b7fde2 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3b7846055fc4f1694e8bfc7a990a158 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8daa61509d7b415a894cf57a00ed8b5b : SUCCESS in 21m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/40a05c29a6284f1a8b6ed1e7981a61d1 : SUCCESS in 24m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6e3891fe54df4347b63888fb83c5c7ce : SUCCESS in 15m 07s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6ae07ba889c54b4b883d443a47e28d66 : SUCCESS in 14m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2b344f34fe24f019ec85253a2dcb840 : SUCCESS in 18m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5e098cd913ef4a45a44ea08882245617 : SUCCESS in 6m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a989cc8c0ac34a81b8f67cb2b123ff60 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbf7ebcf3c1a4f109942d6b682096e08 : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8b447c54cd084e0da6d40abb4b128cbf : SUCCESS in 11m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/40ea4c54b6cc42489ebf80b135762ccf : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84f9f562bf0541cc8997302f0022ec50 : SUCCESS in 1h 03m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1dc71610d40b4f999daed06f5e81d905 : SUCCESS in 1h 23m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ee84b55c5a8423c9ba23541fcb19872 : SUCCESS in 1h 22m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/09bd264aedbe484d9c19ab4acb0eb67f : SUCCESS in 1h 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c263def8759dda9b43d5e8f3a2c231927b42e919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:46:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ea3fea5458a8e3ef4e03ba15ea64b2ff16dfcf1","revisions":{"7ea3fea5458a8e3ef4e03ba15ea64b2ff16dfcf1":{"kind":"REWORK","_number":1,"created":"2020-08-03 14:23:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/87/744487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/744487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/744487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/744487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/744487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/744487/1"}}},"commit":{"parents":[{"commit":"0d16fcc401fd7adaf3bf58a4ba8c636bdbcae631","subject":"Merge \"update constraint for tripleo-common to new release 12.4.1\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d16fcc401fd7adaf3bf58a4ba8c636bdbcae631"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-03 14:23:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-03 14:23:33.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.1.2","message":"update constraint for oslo.messaging to new release 12.1.2\n\nChange-Id: Ibb78aae930bbb98025d806ec4ca9ada04d8ce012\nmeta:version: 12.1.2\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I6870221efeabafaaffd607ee99d6094de799d09d\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ea3fea5458a8e3ef4e03ba15ea64b2ff16dfcf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ea3fea5458a8e3ef4e03ba15ea64b2ff16dfcf1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
