)]}'
{"id":"openstack%2Frequirements~744495","triplet_id":"openstack%2Frequirements~master~I2eeaa6f3ccb3973839dfa83e008f6a3f3a909f8d","project":"openstack/requirements","branch":"master","topic":"add-build-to-gitignnore","hashtags":[],"change_id":"I2eeaa6f3ccb3973839dfa83e008f6a3f3a909f8d","subject":"add build directory to gitignore","status":"MERGED","created":"2020-08-03 14:42:52.000000000","updated":"2020-08-03 20:07:19.000000000","submitted":"2020-08-03 20:05:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744495-1596485140036-e01c6539","meta_rev_id":"6c8eb7d71cdd85698cadcafcecb15cce42a50de6","_number":744495,"virtual_id_number":744495,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 20:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-03 14:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-08-03 14:47:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 14:47:32.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-03 20:05:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6c72d25990ae1776d525b0cefc87c7c0de45228","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-03 14:42:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ea9c041788ad0b5f0c8cfacefa966b01c167e91","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-03 14:47:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nShould be fine, but would be interested in knowing what creates a build/ directory in the root of the repo.","accounts_in_message":[],"_revision_number":1},{"id":"267e8870e28c2409ca1419cf5cc80db01af68979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 18:37:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c28b0790e2b84b9bbdf8f9c41113d9b4 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28cb3315305f48f791abaa2a1363dbbf : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7e70955a3e14fd5ae4db6088e26a27c : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0708ceb025524d5c8bb613388cdb67e8 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4eaea250f2c43b2b44118fe37613210 : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6acfa5c331f64fb483a83fc017d44991 : SUCCESS in 54m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cccc019143b4eaa97e44cb2685be825 : SUCCESS in 1h 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f24807bacc4c45d0ac6b3ce1cd767ef843708688","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 18:37:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"677fd10da7e76d6f6b6a6180b61433349e7df207","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:05:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51458b9d9114f7da033ca67eceb560a : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09ff93c20efb4108b817d7659f71ca6b : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d19e360390542e382a4cf47c27eaafb : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8874ee70ecfb4c42bd26b3781d944249 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f26a057779948b2b7b6c84ee155f366 : SUCCESS in 6m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/570d0ed8f47246b7b42f0b28f8c99245 : SUCCESS in 54m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35f3b743454d4b8ca3be639bc7efc49c : SUCCESS in 1h 21m 29s","accounts_in_message":[],"_revision_number":1},{"id":"815c17644939375eb7e7a5f5111460c541dc216d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:05:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c7d18723b6cdd44a2cc95ff753a7eebf5ae08cfd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:07:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020c94a11f3b400c9f0a7d851008fc16 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54259a621df4e0d9f4a124f5daa4e60dc23bca7e","revisions":{"54259a621df4e0d9f4a124f5daa4e60dc23bca7e":{"kind":"REWORK","_number":1,"created":"2020-08-03 14:42:52.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/95/744495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/95/744495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/744495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/744495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/95/744495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/95/744495/1"}}},"commit":{"parents":[{"commit":"21e13cb12a07f0e1ef285a2a06f7de84abcdadb0","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21e13cb12a07f0e1ef285a2a06f7de84abcdadb0"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-03 14:42:34.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-03 14:42:34.000000000","tz":-300},"subject":"add build directory to gitignore","message":"add build directory to gitignore\n\nChange-Id: I2eeaa6f3ccb3973839dfa83e008f6a3f3a909f8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54259a621df4e0d9f4a124f5daa4e60dc23bca7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54259a621df4e0d9f4a124f5daa4e60dc23bca7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
