)]}'
{"id":"openstack%2Frequirements~744508","triplet_id":"openstack%2Frequirements~master~I029f3173d1de63697131c5c017084500f98b7420","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I029f3173d1de63697131c5c017084500f98b7420","subject":"update constraint for python-karborclient to new release 2.1.0","status":"MERGED","created":"2020-08-03 15:25:33.000000000","updated":"2020-08-03 21:51:10.000000000","submitted":"2020-08-03 21:51:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744508-1596491470834-c0251aed","meta_rev_id":"3d80705d19a0a40e3199bb33014310a4ce04ab3f","_number":744508,"virtual_id_number":744508,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 21:51:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-03 15:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-03 20:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-03 20:15:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 15:28:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-03 20:15:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-03 21:51:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d541cb711e888d319f52339c5195b7405abb632","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-03 15:25:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a326417f45dfbdb880e2f99bb32f8efd33e774e7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-03 15:28:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9cef6b2b9bdcb935f4ce466feef72acae05e358","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:49:22.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a1357aae427d4c01ae5eeeb39ab46ed8 : RETRY_LIMIT in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6e0296271d3de476f26d46d1a41acfbe94e660be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 19:19:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0821d1e821eb4acf97f5b0241d0c52af : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a913ed7167448b922750cd4b7599de : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd7eb798190047ea9675dc33f5d38a8d : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4c34c95d1aa4b9dbd3a6da253f8b5a3 : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de7b27087ce44f039eaf67de1b5bdf7a : SUCCESS in 6m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f7f245ec25c469fb0343d8071a3c021 : SUCCESS in 8m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c85249ca9476490193e7c30bb2375756 : SUCCESS in 8m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5d9765aa52ce4795a533882da90b6a70 : SUCCESS in 13m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3f8798e8256e405c98af476d1f63dc9a : SUCCESS in 12m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ba01b1637f47449895cdc4a726a68df8 : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3dcb135fb6db4787a1cb92b2b791c7cc : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c89f18d4e29446db7602182d9319d16 : SUCCESS in 9m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/486cd4f8837d42db94f1a9a2eea9202a : SUCCESS in 11m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0886921c9aac47b7bac23cb0a30f5960 : SUCCESS in 15m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6ab7800252cc4ca4911f2fb4d60b4f3c : SUCCESS in 30m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/05a2ea242b934a7e9bf5e58d21c6ee80 : SUCCESS in 15m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b7a6e9e3868545f9a6c80abc913ea2d0 : SUCCESS in 21m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6d341697b46641ec86e53a7e26d77835 : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bd32f88ffad74952b778e3d06c501778 : SUCCESS in 9m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/88ddfe305b314f8b995fb9004e9f6ac0 : SUCCESS in 13m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d38dd2043d14e94a6f87dd7f82c8908 : SUCCESS in 49m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b308352bf944adf9731e2d3558d76a2 : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"358f962b52add8593309d027f087bee24f4f6c8a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-03 20:15:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c58218e7cc20bfa927c9d3ad3f10f898a826792","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:15:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"079a2b72b05919314de1240bee4faf2f4f530db0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:51:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66085d923ed543aea3fb46d343797723 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a6d2442df340a8bec1e66812e116b5 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/061d93cb14f84127826db2c6d526c9fe : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b22c99270260469094468c4171cee3d8 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fabe342c42a471b858bc1ceae06a8dd : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8944143e61254baeae6c896167b90b46 : SUCCESS in 4m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1eb6f3f8dba64ca9bffe34cbc32853fc : SUCCESS in 7m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/21897c566620432298faa20094d1422b : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e7aef426fc2e4c5abf300eda4c3a4899 : SUCCESS in 8m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0c26aae6a4234dbab5afb0b7aa21db3f : SUCCESS in 5m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2520d20a077e4bd183a558676acecf85 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/737e966793924304a014aa4c6dbbb5b8 : SUCCESS in 6m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f5175b812eef475e8a1a954a23071cf2 : SUCCESS in 7m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/90b078036dfd47c99d87eb48488879b1 : SUCCESS in 13m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d0b27a6590fd4fad9366a094f9601240 : SUCCESS in 24m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/78916e2f499a41d1bb40a4cec483957f : SUCCESS in 13m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b5112e19c0ad459b9353ac798dffaf79 : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d63499b4367f4321a152311ebafaac91 : SUCCESS in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/157cc62da85b4ebc82ca75fce104691c : SUCCESS in 7m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/da126725cf2343d7b2940ded3249ad7b : SUCCESS in 12m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aae29a79b1c8401099ce1123a621e84a : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd3fa7dc4af24609b811cd1f0ba7f5f2 : SUCCESS in 53m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/464960533c034dcd871105c9cae64f92 : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"81caf7f83c74f653ce2bc4d312fd4eb504c7198d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:51:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd0eb3827317e4de8d22f3e9ad3ca62e25bed972","revisions":{"bd0eb3827317e4de8d22f3e9ad3ca62e25bed972":{"kind":"REWORK","_number":1,"created":"2020-08-03 15:25:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/744508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/744508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/744508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/744508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/744508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/744508/1"}}},"commit":{"parents":[{"commit":"21e13cb12a07f0e1ef285a2a06f7de84abcdadb0","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21e13cb12a07f0e1ef285a2a06f7de84abcdadb0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-03 15:25:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-03 15:25:32.000000000","tz":0},"subject":"update constraint for python-karborclient to new release 2.1.0","message":"update constraint for python-karborclient to new release 2.1.0\n\nChange-Id: I029f3173d1de63697131c5c017084500f98b7420\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: If643d49976e8270d8c8d1653e1a3055919eff21f\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd0eb3827317e4de8d22f3e9ad3ca62e25bed972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd0eb3827317e4de8d22f3e9ad3ca62e25bed972"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
