)]}'
{"id":"openstack%2Frequirements~744509","triplet_id":"openstack%2Frequirements~master~I7388f57fe53a0026bb12f41926432bd5276194ce","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7388f57fe53a0026bb12f41926432bd5276194ce","subject":"update constraint for cliff to new release 3.4.0","status":"MERGED","created":"2020-08-03 15:26:03.000000000","updated":"2020-08-03 21:51:12.000000000","submitted":"2020-08-03 21:51:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744509-1596491472270-3eeecb9b","meta_rev_id":"7ce4b259d43b64ba482efba15f0cb6587281e2bf","_number":744509,"virtual_id_number":744509,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 21:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-03 15:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-03 20:15:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-03 20:15:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 15:28:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-03 20:15:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-03 21:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93806ff45130cafb8adce27aa78e23ff9a93ff98","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-03 15:26:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b3d7f951443317e5bd02c823f9fc8f2805ffa49","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-03 15:28:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d617126d4768d227b397b3aa5cee58141be5d99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:49:10.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1a5ab2a8f3f64f77abcac47cf92d1038 : RETRY_LIMIT in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"4ce26560705bd2ddd80deefad32457a80909d02e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 19:48:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8933d2aae37c4b32a9957bfeb9eef2bb : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2181c7f24c474a9fb93c9a6c875f9e90 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01d9f4d8dda9461f88e7026aca3c74ec : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c07ee1d3b7d4756bdf669d7b3483a56 : SUCCESS in 9m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50d08c6d812d4d4a8df3172f8d24450f : SUCCESS in 6m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8fa42005ec56428eb770404a62fa9f1c : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6681a3b1e566450aa51df46c6ecd5fab : SUCCESS in 8m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/30c1770ae08a45a695a39386d8b9433a : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/db798af7931e42e8af229e7d1a4e592a : SUCCESS in 11m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d29df712920444dd949fd2fe5355df6e : SUCCESS in 6m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/786d6f5bbc524888800de961782976fe : SUCCESS in 15m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/941f4b39efda4e90b80d686eefd632fd : SUCCESS in 9m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/36a11132a75242bc82e3f7c9462f37a9 : SUCCESS in 10m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0bdd22d950794ceaac805fef69ad655c : SUCCESS in 15m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/321b820d9a30421bb29f1a21ff9b3745 : SUCCESS in 28m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6bbcabc59c2647cb96c6521801c38263 : SUCCESS in 17m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cdbee9a2d26f43368a852423206b2d01 : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/547a7c7f62e34f2f8fd96db456482304 : SUCCESS in 7m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a92117854af4ce7a44a0862bc2670a3 : SUCCESS in 7m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/81f3951d8e0147d7b7846f8c48d27946 : SUCCESS in 14m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df492395ad404a36b4a53c0fcadb0de4 : SUCCESS in 51m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d92b6999d3a4d2ba74231d3836ca990 : SUCCESS in 1h 43m 57s","accounts_in_message":[],"_revision_number":1},{"id":"47f1fa301552aed0540c866cad4ba05200a02d94","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-03 20:15:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1141e14003f773047423cba10c6b9b0641d0aef2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:15:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a426de0408f7aaa8e3f9ece0cbf4ec454905793c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:51:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af3ac71d99594381974d0e7b129531b9 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4559695ae7084fd99d9ad2847efd572e : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78ee7badfd9c4ab4afd3b8d93a8a1a4c : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2b84491867e473c9b42f6b583a13d4d : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96b2aa36a1ca4baf9129278c15cc77b2 : SUCCESS in 3m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e5f65a13d6144bea87ef36f29774032d : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6623458c434d4b4caea8eff6c882153a : SUCCESS in 7m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83494ae6641e4c59b76bde1f51500495 : SUCCESS in 10m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f8c39438479488c9a20c2074913a04f : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab12ac3d364f459aba67222be0156bde : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c54bbbe7b894b5da4c72239fa50cc82 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/620dfb7d73424c5a928b71850535f80c : SUCCESS in 7m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1d5a4ffef0eb4b989eec2ed08dc4c6e3 : SUCCESS in 7m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6093fd6f4dd54254aa1a70b1de9616f0 : SUCCESS in 16m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3343cdc170084f19b7a881601c8f9a5c : SUCCESS in 31m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dcec6c8bd9fd4432a8b2d253e207345d : SUCCESS in 12m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/700bd65223aa44c39e14b659563ec9ad : SUCCESS in 18m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/97e8acb4fc0b41a29c484e1298e4df37 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/27a9270b2b01431ba645e56f0a2e2fdb : SUCCESS in 7m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/51a5ef9a768a470a86c4ca2f20dcfd11 : SUCCESS in 10m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0edb0a2ba0dc48708a2a00cd2b9485bd : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40706bc62ac14b6db67f26f07e92e7da : SUCCESS in 1h 02m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/710f74c9eed74d1d8bf1849b91ca36c7 : SUCCESS in 1h 22m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c25960e9254b671bb8ab5c35e3501b70af9d547d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:51:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"115cf4b8528e932099739abed9177b606cadc91d","revisions":{"115cf4b8528e932099739abed9177b606cadc91d":{"kind":"REWORK","_number":1,"created":"2020-08-03 15:26:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/744509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/744509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/744509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/744509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/744509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/744509/1"}}},"commit":{"parents":[{"commit":"21e13cb12a07f0e1ef285a2a06f7de84abcdadb0","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21e13cb12a07f0e1ef285a2a06f7de84abcdadb0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-03 15:25:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-03 15:25:58.000000000","tz":0},"subject":"update constraint for cliff to new release 3.4.0","message":"update constraint for cliff to new release 3.4.0\n\nChange-Id: I7388f57fe53a0026bb12f41926432bd5276194ce\nmeta:version: 3.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: Ib448ebc0ee920677ff1a33cf6dbca6f86e716c54\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/115cf4b8528e932099739abed9177b606cadc91d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/115cf4b8528e932099739abed9177b606cadc91d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
