)]}'
{"id":"openstack%2Frequirements~744510","triplet_id":"openstack%2Frequirements~master~I78878cdc58ba929bd406cebae65992db65918d08","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I78878cdc58ba929bd406cebae65992db65918d08","subject":"update constraint for python-designateclient to new release 4.1.0","status":"MERGED","created":"2020-08-03 15:26:17.000000000","updated":"2020-08-03 21:51:13.000000000","submitted":"2020-08-03 21:51:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744510-1596491473919-ff0595dd","meta_rev_id":"4b9cc87c6c680c1be6b24695dacfbf853ce9d6fa","_number":744510,"virtual_id_number":744510,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-03 21:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-03 15:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-03 20:15:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-03 20:15:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 15:28:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-03 20:15:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-03 21:51:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c53eed190465de80418144b51130bb9cc741353b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-03 15:26:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"343ffed84b504ab6a0f0ac3184f64a7496e82737","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-03 15:28:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a6bc1336315659551e59550c95179d001590595f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 15:48:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1dd6c06137c14ec9a0b55dedb6a81e37 : RETRY_LIMIT in 2m 23s","accounts_in_message":[],"_revision_number":1},{"id":"dba313ae51a10da2eb6476ccf67ba84570d76de8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 19:40:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91f25e492a84910acbdbca812302f6f : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6026cef1811643a4b74ce62c92c8ac0b : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7c98aa58c5343a58c4c3850bf1119bd : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f646c07b83f4553826300922363d6c5 : SUCCESS in 9m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a531311e8cbb426cb6ec0c97b9d8de78 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f70d91d8b5ec4d40a33efe87954ef455 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3f1fdbbbd12e459089ebd9cfc6b1fbd8 : SUCCESS in 8m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d7f4eac8e5d446a5b85ddd94da630620 : SUCCESS in 12m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2bc5d06c11444f769d4fc1575257fd5f : SUCCESS in 11m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4437741d104f4261a6b77cf3ff6aa117 : SUCCESS in 11m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d687f454e7d94a608f00412191304933 : SUCCESS in 13m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19578a19151b4dd4a416e93edfd4e49f : SUCCESS in 11m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cc7b4afbdaef4ce297bbd6193337bdbf : SUCCESS in 10m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dad638f3d2084831b6ca9a9a39e02935 : SUCCESS in 16m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/29b09a5e95e94216b98457d0eacb1730 : SUCCESS in 30m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/34a28cfb325c4b6cbb99e783d8d7a0e9 : SUCCESS in 16m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ba64b58549a4fb7884ef97b7cd11651 : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f3848a150e88481dbb65ea6d2b6c66af : SUCCESS in 9m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/931cd25868bb44ee9de78f46c263a450 : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1eb73c5957a1421e83e091f813c109d8 : SUCCESS in 17m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48cf4062e02a41d3b440d142be383269 : SUCCESS in 51m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b25967c8fbb74900a6cf383584cb1d0b : SUCCESS in 1h 30m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6eebccc578008d18998cc11a3d32b030759e4bfa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-03 20:15:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9da6d6de45bd2ee94a9dd646abbd9761a1414fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 20:15:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f94f4483125f00ab9ef02d279a649339c47fed67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:51:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8b5ca8fe9941c09dad58e71cdff298 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c63c0b0ac84453918ef2ec46c69c50 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caf6fd2e25ea46969f2dcecc20833c00 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb25c55c9aab429fa47a5885e803c770 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e9563d881824033889bb12e0c3311eb : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/020d904a5c82411bb7833ad0d362d1a6 : SUCCESS in 6m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce9b3e177f3945a5962683af904cd896 : SUCCESS in 8m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ea5cf19fe1b4453b9faac21c762bb49e : SUCCESS in 9m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4efb35159bc942549d7faa364d5a3b73 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cb1a8411a3414ef9910013360e979ffa : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b32a9dbe671249319cf3ab70d9d858d7 : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58dc3beb0908475290d1014aba1ab39a : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3e554b0dd24c4408817c23859c1ad2fd : SUCCESS in 8m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1d97f95fc0ba47d9afce77ca063e2898 : SUCCESS in 13m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/85960528c5c248e4870f610c25047d99 : SUCCESS in 29m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/232966a7f8de4892bb6ea7db4f6d57d0 : SUCCESS in 13m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ec6479c1d72a458dafcc2d2f2bac277a : SUCCESS in 16m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c698f4076c724c7aa3a1e0b162e76845 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/280cdd0d512048c2865cfddd40dba79d : SUCCESS in 8m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/af68d0de7bed44d8832087daeac7b2c6 : SUCCESS in 11m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9182b34c176e4e2ab0e28cc8426ac0d2 : SUCCESS in 12m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/100f14c078f342d0abf941ac0918efe2 : SUCCESS in 56m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/faf375fb390240cfac06e753ee80117a : SUCCESS in 1h 21m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c7de7795f532d518a8666d57a00030ed208b26d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 21:51:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ea8341ff9ce99f8f81deb28718602c0f5e7b289","revisions":{"6ea8341ff9ce99f8f81deb28718602c0f5e7b289":{"kind":"REWORK","_number":1,"created":"2020-08-03 15:26:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/744510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/744510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/744510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/744510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/744510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/744510/1"}}},"commit":{"parents":[{"commit":"21e13cb12a07f0e1ef285a2a06f7de84abcdadb0","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21e13cb12a07f0e1ef285a2a06f7de84abcdadb0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-03 15:26:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-03 15:26:13.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 4.1.0","message":"update constraint for python-designateclient to new release 4.1.0\n\nChange-Id: I78878cdc58ba929bd406cebae65992db65918d08\nmeta:version: 4.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: I9329bc7cd2a3e3c80f25fea1b54ae19905a2396c\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ea8341ff9ce99f8f81deb28718602c0f5e7b289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ea8341ff9ce99f8f81deb28718602c0f5e7b289"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
