)]}'
{"id":"openstack%2Frequirements~744649","triplet_id":"openstack%2Frequirements~master~Ie9aa22915f3c3c6da74095da6db4cf01e5885bb3","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ie9aa22915f3c3c6da74095da6db4cf01e5885bb3","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-08-04 06:19:40.000000000","updated":"2020-08-05 22:55:21.000000000","submitted":"2020-08-05 22:54:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744649-1596668045594-c4c333d3","meta_rev_id":"67ec4edd4f7ea785f6c049e570e439a4a9c10b5b","_number":744649,"virtual_id_number":744649,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-05 22:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-08-05 15:48:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"date":"2020-08-05 15:36:33.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-08-05 15:48:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 15:36:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-05 15:48:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-05 22:54:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ba4f683912cef4361e5648a0227a44690cbd482","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-04 06:19:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f895e1f98471c552359cfdd54d53d434a27f4665","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 06:54:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/145d281342114c8eb6ba00d64da332b8 : FAILURE in 19m 34s","accounts_in_message":[],"_revision_number":1},{"id":"957097234b76fdfca83161a4fe16898b54303fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 08:34:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ee02cc72da4f68b14cc172ed5d281f : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e5976319514f6eb4f93ddeff17c037 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e62319dedde843f994856c25cd7487b8 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec0de82463254aa599abbbe30d4175be : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fff68d6b0f23426b963bb2709167f605 : SUCCESS in 5m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/346b91326d43439aa7e3e4455b00a6fa : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/329aaf3a912f4dbc9b4339bdefb95bd3 : FAILURE in 8m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/32afd329131b470ca9cf44ba35dddb2c : FAILURE in 10m 24s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ad6b45858d36449d8ae12d44806211e0 : SUCCESS in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f85a1182426948cba2d5b84fa7f831b4 : SUCCESS in 6m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ee8ae7a77c91487991974b07932bd696 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/13f3570173cf4112b81b1c21fe1744fb : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ce07801b26e487faf4cc48a40881abb : SUCCESS in 8m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d95ba0e6679b44538d3fdf2eac842506 : FAILURE in 21m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d99ac3cd7914c83b9be7491d8c6dfb9 : SUCCESS in 22m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5c0aaeb714be4973b676884d3f13d409 : FAILURE in 16m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8d03ee148e464997938321070a84b970 : SUCCESS in 18m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c84ad45dd2044780848e5d5b79c8437e : SUCCESS in 7m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb1110e9ca3f468293de9a074670e962 : SUCCESS in 9m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5caa1b1a93f848ed8a6251389741328a : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2505140fb0534e3bb6e7fa91c42ba22b : POST_FAILURE in 51m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e30f2b93147f4a55a6d18caf73583db4 : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":1},{"id":"53580e689d5aacfe73f002ca2235df38b37c65b8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-05 06:16:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd4128e357ab302fad5aef07963f744171a6799b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 06:45:08.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c5c894e2c48048be9e30822e868143c3 : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":2},{"id":"3d27241d7cc302a395198fffe061eee41e1e732a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 08:31:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94fc1d42e13b4caf81c70527f5df2f46 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1251688a219a4bf2bcb80ea106a908db : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbe39c18f59049ebb0ccc74d033788f2 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d4723eba27049e682290dcbc32d4c3b : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4af39977ac3649e4ad1cfa06e4383e91 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9222773675c44896b5f17aa56ea6f52d : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/21e2378fab3949ec82f1251b33bdfd9f : FAILURE in 5m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e212891661124e91bddb989d235935b4 : FAILURE in 5m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/378d7446b2fd4111aa4048c7bb38fd4c : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/26a4ca7fd0f14a898cb0bb72504d52c8 : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53735b9312f04085a62c9900a7806ae8 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44aec6a76c1e4e3293082161aa1b0e95 : SUCCESS in 6m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0ff294c5434e4769bd4576807fede8b6 : SUCCESS in 7m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/05f6159f95564d51a2fb75f86b759997 : FAILURE in 11m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3065366f7420485bb14fa6442f5ff158 : SUCCESS in 16m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f6b8f583cebf40358db82bab2c41452f : FAILURE in 10m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e14f7b7cd40b4ba685e0fb3cf2308e6f : SUCCESS in 12m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8005e734cd4441cacf3a5117620b4f5 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/42443e68b46f4b2da67c7feb8765b704 : SUCCESS in 6m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ce6d1642613c473f9032fdf3e6d2b87e : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51507dfc68af4aaebccefd5a3c2f923d : SUCCESS in 1h 09m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/690c3c1266c14ce0a3a110ec7b52c139 : FAILURE in 1h 07m 59s","accounts_in_message":[],"_revision_number":2},{"id":"9ca8f2894c2d62ce30c450df8715e73270be0664","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-05 15:36:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a6de8ed5091c97cd1139227b4a9f3551beb5774","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-05 15:36:40.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"5d4ba659fbef83d2ecbda87501c2a519fdcb2e77","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 15:48:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"30b3f1c6051dadb0896712f989bc4b2ad7f9675f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 15:52:55.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aeb979b58f5a43ffb86da17af35e8f8a : FAILURE in 12m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ad3e5eb78a4fcec809413370a790d565b1772c90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 19:33:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e6aa08c5b5476aa02829247da33286 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0afb3651606a46d08a77893c647a4b66 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65d6ce663ee7461badddee9d4477dcb0 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d250898303a4f09868cfbd1f8d966bf : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4694e0fa67534e988aefc02d836ee563 : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7a95691f5934cefb4630da771358297 : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aee836d1a97b4ad8a8f6b712dae09556 : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c6f8341e20144692a4bb40d84353f0cb : SUCCESS in 10m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f9ac27817b7b46989a835dc69093b19f : SUCCESS in 10m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5130d896d61b410b94ee4cf77c41e7ae : SUCCESS in 5m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ba5fbf23820b4fb8be9d92056601eebb : SUCCESS in 8m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2c9a5cc57a2e428bb521bfd0cea3a3de : SUCCESS in 12m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d4e9b8b85699439c8efa0c6f59bc79d4 : SUCCESS in 14m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b8e265a13bd34e62b8b336ab38f6abb6 : SUCCESS in 14m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/47f53760f0d2479fa15c68db6cb77270 : SUCCESS in 27m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d63724827e6742daa1e487c8a37bf71b : SUCCESS in 15m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/082ebb1af93442089db235201c172853 : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d79f617f82a84626832bd49d9a4709bf : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04f164a104ba49bd9a2399348b66d270 : SUCCESS in 9m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d6286634bb104faab5b2b6bba1180329 : SUCCESS in 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b902174c55c426a9a43757e9c3273c4 : SUCCESS in 1h 04m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc5631730eb84f03a5f57764f91ca802 : FAILURE in 1h 13m 41s","accounts_in_message":[],"_revision_number":3},{"id":"d6b77fa1c06210c389aeca46990f5b64f7f75ba6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 19:35:46.000000000","message":"Patch Set 3:\n\nrecheck\n\ntempest.lib.exceptions.SSHTimeout: Connection to the 172.24.5.100 via SSH timed out.","accounts_in_message":[],"_revision_number":3},{"id":"e4a93c8239bcffd7b14309e7595fd0fe4d4b7a96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 19:55:16.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f5b1e62045274a4385f5301009094437 : FAILURE in 12m 10s","accounts_in_message":[],"_revision_number":3},{"id":"466af8a6041aab3535fb409053ef11e9d70f53f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 21:04:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb2f2b1605f640f2bb4920632e7af469 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f43eecdadd49d09442c480b0c1ccdd : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fb0b1801e8643abbaa703d0710e985c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c727f4eb5e148c4bda5f5b6647d9316 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e2dd6575c79419eaff542d39f93da0a : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a91e924d7fc4847a3f9db02ce967553 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/55d18838075f4d04ac1c6ddac6a641ce : SUCCESS in 6m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c062c7cc56f346d1a3857b5776822d9c : SUCCESS in 12m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/33a8f6ea0319430c8241361063414b25 : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/afa43d6fb5284eba903b5336e6e73c09 : SUCCESS in 5m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d434a492477d4f5983dbace5e0d450a3 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5703fbba8abe4011bb91e635329c41a9 : SUCCESS in 8m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5bc76f624efd456fb87e4356ad258678 : SUCCESS in 8m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fa3e267e96f34943882e24e386890ae0 : SUCCESS in 14m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/73b32b35f7bb4a0085f10d6877fe9cad : SUCCESS in 26m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bae175b885d649dab6fe846b75e5b889 : SUCCESS in 12m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ccf85c10c05f4b318dbbe81c005baa15 : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/42edc39c7f584b44b421cf767031565e : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a7b858420dae429bb83995e9bf922a4d : SUCCESS in 7m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/46dc822b70f84a14983bd98536a8dd3b : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/294b455da9c246438184a43df6236a42 : SUCCESS in 1h 04m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b5e0a8db75041e694f8df3f19f81186 : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":3},{"id":"7cb8d4b3c80e0c1ae35dc95778f2a37a3c73dbe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 21:05:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"021a29c759b0a56a81ad4cf7ce8e410c3fe09786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 22:54:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca4f99515f7746b78770fc86da3c1602 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5952919863d4e99878e5631e80456e5 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b559629e463c41b4aeaa1e432b850794 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7541b1e5231247bdad5db72e5c2b0742 : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ac5725454b545f5a5da578c2d5adb1f : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4b971fa894e4cc1ad44bd1d7898f250 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3b0e9284dca04aa68008a45c736018bb : SUCCESS in 7m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/590d70c405d3484cb6023b22f2746762 : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/71a54d637d67455bb824ca87a33737fb : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f314b2bc2ce84e43899cf4c3a36e82dc : SUCCESS in 6m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/13c9e43768084c199aa557746aa5ebfd : SUCCESS in 7m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b19ef0f7952b44a8b41a3b1134f69b69 : SUCCESS in 7m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b3b4945490634fc6bea066dd022abe72 : SUCCESS in 7m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a7811f956a5043c792c2173685fb929a : SUCCESS in 14m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d5d7f0a20879426e9ffcf30311fc261a : SUCCESS in 27m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a9fcb57a88174d899c810529cd123a88 : SUCCESS in 13m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d2c85ded94334f8ab2b36706b3ed2907 : SUCCESS in 12m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c4c7cf20b6544142bb6be17f81fa65e5 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec555d41dbeb4fe5b4a9ace0d1e3a774 : SUCCESS in 8m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5122bfb7894547ba8fd9dff8c5371651 : SUCCESS in 10m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fddedcaaa9904c45aa4c1a467feaccac : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e69eb1c179694fe69d86eed6223fda1f : SUCCESS in 59m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9af81fc3ab5249eb87485536c0b08a53 : SUCCESS in 1h 26m 06s","accounts_in_message":[],"_revision_number":3},{"id":"4647de9d2034062f59563726e9acc258c137451a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 22:54:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"67ec4edd4f7ea785f6c049e570e439a4a9c10b5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 22:55:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68df1ae58d51440f910afa1922d505ec : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7dba200f2a3e6b56dfbbc4e8067db905ff82aedb","revisions":{"8f27a397e01520c9a393a5331e5507ca9ba8a669":{"kind":"REWORK","_number":1,"created":"2020-08-04 06:19:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/744649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/744649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/744649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/744649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/744649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/744649/1"}}},"commit":{"parents":[{"commit":"ba5858c4d1c46e3da619deb6a4cb79064084b76f","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba5858c4d1c46e3da619deb6a4cb79064084b76f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-04 06:19:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-04 06:19:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie9aa22915f3c3c6da74095da6db4cf01e5885bb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f27a397e01520c9a393a5331e5507ca9ba8a669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8f27a397e01520c9a393a5331e5507ca9ba8a669"}]},"branch":"refs/heads/master"},"7d3eb6e31894c27412e7285045d8b7704bd8a240":{"kind":"REWORK","_number":2,"created":"2020-08-05 06:16:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/49/744649/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/744649/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/744649/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/744649/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/744649/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/744649/2"}}},"commit":{"parents":[{"commit":"a44dfcc12233e940d7d2b9b0f853b49e83d7579b","subject":"update constraint for keystoneauth1 to new release 4.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a44dfcc12233e940d7d2b9b0f853b49e83d7579b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-05 06:16:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-05 06:16:05.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie9aa22915f3c3c6da74095da6db4cf01e5885bb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d3eb6e31894c27412e7285045d8b7704bd8a240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d3eb6e31894c27412e7285045d8b7704bd8a240"}]},"branch":"refs/heads/master"},"7dba200f2a3e6b56dfbbc4e8067db905ff82aedb":{"kind":"REWORK","_number":3,"created":"2020-08-05 15:36:33.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/49/744649/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/744649/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/744649/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/744649/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/744649/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/744649/3"}}},"commit":{"parents":[{"commit":"a44dfcc12233e940d7d2b9b0f853b49e83d7579b","subject":"update constraint for keystoneauth1 to new release 4.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a44dfcc12233e940d7d2b9b0f853b49e83d7579b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-05 06:16:05.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-05 15:36:19.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie9aa22915f3c3c6da74095da6db4cf01e5885bb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7dba200f2a3e6b56dfbbc4e8067db905ff82aedb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7dba200f2a3e6b56dfbbc4e8067db905ff82aedb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
