)]}'
{"id":"openstack%2Frequirements~744666","triplet_id":"openstack%2Frequirements~master~Iea372ce034acc1f4b95c09005dde28e1537e7b7d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iea372ce034acc1f4b95c09005dde28e1537e7b7d","subject":"update constraint for keystoneauth1 to new release 4.2.1","status":"MERGED","created":"2020-08-04 09:23:23.000000000","updated":"2020-08-04 17:27:50.000000000","submitted":"2020-08-04 17:26:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744666-1596561969931-5080a3ed","meta_rev_id":"ad85999b04529a2d129c0b3e214a924a6daaf700","_number":744666,"virtual_id_number":744666,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-04 17:26:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-04 12:12:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-04 15:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-04 09:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-04 15:38:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 09:58:24.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-04 12:12:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-04 15:38:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-04 17:26:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b95bf12e1a823a349f5b991b34d92458f6a94104","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-04 09:23:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7471ef26102f5f72f9bd710f0b05d495849ffa49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 09:54:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2647fd8fa0364b92b2b7ee87923d2e14 : FAILURE in 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"685205327bde777ee11762f9d3055d9455308734","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-04 09:58:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3879aba8682151a64038bc0288b18c3f882d2e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 10:56:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4e13b6b49c450893cf8f7294614bc8 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b66e2689e0134d4d9da9da61268fb4b3 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b344bc45c394e7d8900cc35a5e01b32 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f3c45c33ab14e499698a4e63b4ab551 : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1a534c0251ec45e4bd68f8f567a35050 : SUCCESS in 5m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97ec74ffedc340ac9209b19dfbe614cc : SUCCESS in 4m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/95d5a3d26bf74567b4bd15f9b70a8708 : FAILURE in 5m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1853fcffa93442e788f2220ac7637b7c : FAILURE in 7m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d8a03b6edf534f5cb1944b518d865899 : FAILURE in 5m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d794f6af25094a008294ffd4fc4f6abf : FAILURE in 3m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d944882fd2004a2bb280d8fbf96273c8 : FAILURE in 3m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1fe739f22d3841a1a8025a19bbc49f9f : FAILURE in 4m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ab52ff39d0924b57aab3379d2595ce40 : SUCCESS in 12m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1fe00cc9b072438ea81de39e4ddc0676 : FAILURE in 3m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3fbdfd1207eb49948f064e447ab7f93b : FAILURE in 3m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8d908217465f413eb9476d34a1fa1df0 : FAILURE in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/01e5a4f8119a42e3af1d319c7b34a049 : SUCCESS in 15m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2110b5448a384c75a0b50c6f78cc2d20 : FAILURE in 3m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6fc7f3448ea943168cf287d479c6ff69 : SUCCESS in 6m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ff4d92ee9a0b4be8849dc5c265d52cf5 : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f9b887134da45deb58be54e8df6b285 : SUCCESS in 56m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4b134cd87234528bced88d24e1a7968 : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d215af307e539f20e1c9fe8b536f86cd63a3b4ec","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-04 11:07:51.000000000","message":"Patch Set 1:\n\nrecheck weird, the new version seems to be on pypi and tarballs.o.o","accounts_in_message":[],"_revision_number":1},{"id":"2638ab572ed2ba1266f68378cc0c67952e271d4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 11:21:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4964c6eaef53448e8b0e9da6481de2b4 : FAILURE in 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b2373dd4449888c2737649f0fb7c51e88e661b2d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-04 12:12:50.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nLooks like jobs ran too fast and the package wasn\u0027t available yet.","accounts_in_message":[],"_revision_number":1},{"id":"4dd919478b87958c3dda1a4d10b479c084700fdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 12:30:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4565e7de4dd440a2a59a317bde3cee52 : FAILURE in 13m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d856a39f588d0e77bb32edb1f41863fc878d461e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 12:32:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4adb51e0d1df4757aae0f453772de6e4 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a358451021e49f08530edd92b9ac703 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/297ddd1305824cf490febd3e0fe0f86d : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c13f1cbf46348c78ed88883e2d29816 : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f1a2b1ef0e5e4678bada721e2ee5fdfb : SUCCESS in 5m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ada8d64a0344165bc1f5196eef213da : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a60db06439f446fd84a63b725f2ca695 : SUCCESS in 8m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c49f6c918224825b1ca4babe5aa9fa7 : SUCCESS in 10m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb3371f349ce4d78a19fc6e92827827f : SUCCESS in 9m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/435412a34ddf417f9f1d54f30232e755 : SUCCESS in 5m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/61f27bd88ec14a35983c2907420bf5bd : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0acc33dc921425a8ba1ae2ac2dc2920 : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c9dfa68764fd47189198b27e19971c2b : SUCCESS in 8m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ef5c43899a2c4343a0d9b2b4f1823b3b : SUCCESS in 19m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/11b26e1d95574b6393a9ce980b67df99 : SUCCESS in 21m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fc1ae15abd1643e4a3260f4a325c9ecb : SUCCESS in 16m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9ab25277341417b9eccd6ee6183bc6b : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d1f34df85f40431d84907787be646484 : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10433727505c4fe3ad75621d539c948b : SUCCESS in 6m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cba0d8c0b1c941b2bc1c6bd55f2671f2 : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f74b8cc0ee61460baee74101f0d2486d : SUCCESS in 1h 00m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e60f307923a7416f930c0394ceb47beb : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"92ab6162324b802f2c1d2628bd5be802e7bc62eb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-04 15:38:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9b3e4454b8d46347b7233e6c19e3695d9c1c372a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:40:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e1c5dc2424e2e2c51403303c2c741f377a383f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 17:26:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"df82e1e54ade0c784c4b7789394cdf538c1de7bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 17:26:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6dd243277624236a290a79930562ea7 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd892da6df74478c94fb3ca74f0397c3 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d96030ae6de04d7f95b3d4111411f46b : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ce4a12e7cec445099fb7df87291dc38 : SUCCESS in 6m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03e6ccd397134bd3b632c3d880fa94bf : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4ed711bb1e124bb3ab077b562c8d4f38 : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f10e099aa4ac4c94a30525f0b1823d83 : SUCCESS in 8m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a1ea6897a0c64c3b94c604fb877cafce : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e07f844f1d1f4e05a5141e1d0f3b3e24 : SUCCESS in 11m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/08547812d5fd480fbed35bdbbf8dae3c : SUCCESS in 7m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d06603088174442a96ee042bdb12dc4 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2cbf6ef50ecc4830b843985d66de0c48 : SUCCESS in 10m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/94b3d49d04ce4468bad4d67519222d3e : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2e82065b2495408b97d06336f3b28d28 : SUCCESS in 15m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f2a66b25cfb940b59ad18685308cbd22 : SUCCESS in 27m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fadc1f70319a4d8c84017318ba9f6754 : SUCCESS in 16m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/259013b6766548418c4d68f9362d8d1a : SUCCESS in 19m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3ff1b26f521d4c689097f1baa5c77d44 : SUCCESS in 6m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ba5c0ad20b04605bd5c3fd51ded3f35 : SUCCESS in 9m 24s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1b634c4c52f546b0b44ddcdbe88e5229 : SUCCESS in 12m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bb27768efd9c48888cecfe8dcedaffab : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03a178f22b4f415d9cdd527aa84c216e : SUCCESS in 1h 03m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a4a28cc310447269b650fb0a120a627 : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":1},{"id":"4dc37213d319db9edb2bfbb8a4467ec99ee1acb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 17:27:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8bc6c0d6c9449c9940403f6d6c861d3 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a44dfcc12233e940d7d2b9b0f853b49e83d7579b","revisions":{"a44dfcc12233e940d7d2b9b0f853b49e83d7579b":{"kind":"REWORK","_number":1,"created":"2020-08-04 09:23:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/744666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/744666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/744666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/744666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/744666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/744666/1"}}},"commit":{"parents":[{"commit":"ba5858c4d1c46e3da619deb6a4cb79064084b76f","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba5858c4d1c46e3da619deb6a4cb79064084b76f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-04 09:23:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-04 09:23:20.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 4.2.1","message":"update constraint for keystoneauth1 to new release 4.2.1\n\nChange-Id: Iea372ce034acc1f4b95c09005dde28e1537e7b7d\nmeta:version: 4.2.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: Id6171dd2c9eea97d8e44234da8175cef2ad94598\nmeta:release:Code-Review+1: Kristi Nikolla \u003cknikolla@icloud.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a44dfcc12233e940d7d2b9b0f853b49e83d7579b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a44dfcc12233e940d7d2b9b0f853b49e83d7579b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
