)]}'
{"id":"openstack%2Frequirements~744736","triplet_id":"openstack%2Frequirements~master~I278af4d6868273081d7579101320e0c573bd4084","project":"openstack/requirements","branch":"master","topic":"lsscsi","hashtags":[],"change_id":"I278af4d6868273081d7579101320e0c573bd4084","subject":"update constraint for os-brick to new release 3.2.1","status":"MERGED","created":"2020-08-04 15:27:17.000000000","updated":"2020-08-14 16:57:48.000000000","submitted":"2020-08-14 16:56:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744736-1597424184216-de949bf5","meta_rev_id":"fc10d5035b8c441696b16e930716c3e983c93295","_number":744736,"virtual_id_number":744736,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 16:56:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-14 11:02:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-14 15:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-14 15:41:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 11:02:37.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-14 15:41:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-14 16:56:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aae7e47c5ca4a2a322008b590c25c24d96f6e722","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-04 15:27:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"392fb9a24fc9ed15bdf9ba57d3412b6096640713","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:51:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6d6688b118fd47efbbcfc0ee2b3d7ed5 : FAILURE in 18m 59s","accounts_in_message":[],"_revision_number":1},{"id":"88e0a51723b1d205bc00c7ab08d2c803016dd2a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 17:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e5d65fdb1f442585f14d193ceb1792 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f0f89b3db444824a1e009c761366ed1 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/790a0bb68a1b4a42bc0bdfdd7f8dd52a : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9f7cea806c44e519c590608b056dff5 : SUCCESS in 6m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7000da1fb7644df9888ff24cced1cf59 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3b7242c9d514688a971ff420cbe4434 : SUCCESS in 5m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b798af957e8f4c94843a8de23c648a90 : SUCCESS in 10m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a7b5a57cb86419d92ce1d5296fd6f83 : SUCCESS in 14m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bf75fd19c4c84a618d4eca9007b37bd9 : SUCCESS in 12m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6abe23385c644c05ab3692fae39ee2d9 : SUCCESS in 6m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63c28cbe2ff54f649a88312a20ce369b : SUCCESS in 12m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/340320eef1924c4b84350f873166bd86 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/06c57336a36a49589790a5ee121e1f19 : SUCCESS in 11m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/649c3285b27f47bca994a8786e7d7baf : SUCCESS in 15m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7e584d74b84f44afa4d92f243398013d : SUCCESS in 33m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e8adb4d1d59d4455aa888d2661e264fd : SUCCESS in 13m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3a69a3ad96c842a2977a3089e3bfefe9 : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/21b50e800e964c8e9eafdaf6ec4e02e3 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05500365ce8343a6a67db8c4597ec5fd : SUCCESS in 9m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/69e2f2c887f4436eb5a4134984d5d0c7 : SUCCESS in 15m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c6e597ba3fb42d9a8515cb3f8ed78df : SUCCESS in 51m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3528fc5abdb4a63b08cfdc756de3312 : FAILURE in 1h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5d442196b6626dc3ed00ed6f8c77968115944337","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-04 18:07:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab040b076852f493cdfbc6d615a85536b55219a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:34:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bfdd38b28c1a4f1d8e2b79134063bccf : FAILURE in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cc4071618df9779b4d5a507a92743a4ce82af8aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 19:29:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b48baa3ff4e4894b0cfad95c76a1507 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6b398f972f4411ba4d8b0001ed15c8 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/831778e1abd041aa98764f699819fda0 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/209e2171246f4604994fe16b4fb1b86e : SUCCESS in 7m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/770181c993114d66af9476979c37fe0c : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/70d89ed7c9334d4f8c1d6e5b67d847c1 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6bfca48bf62f4aea8d4dbd522f293f3f : SUCCESS in 8m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3ca9b5e852974d1794361c699fd829e5 : SUCCESS in 10m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6f10266ceab14c798f59081746041133 : SUCCESS in 8m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f68cc206c5d740e9918087a210164f70 : SUCCESS in 5m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/130ea393b18d4e218b779976f06dda0e : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a455c1185994449c908a8ec39e3f79b1 : SUCCESS in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c2ff623b08bb43f6a78cd3dcc43df78f : SUCCESS in 7m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a143f4433c524eda9c246a9dceef44d3 : SUCCESS in 18m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c58c5cd0696f40c0b1e2bb6bc494ce8b : SUCCESS in 25m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6c6a54f3d539487fbe17005d02d2df16 : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b4f423f432834e60a861255a5e3bc501 : SUCCESS in 19m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f33f980fbc2e44229ce2a7d5991dcea2 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f181751c81bb453aa8c295d1e09a9a57 : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/264d4ce5b27e486aa7dc0b6e0c419ecd : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84bf6108404f42cd815eff83da10a858 : SUCCESS in 56m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e83846e6a7d54eeba2801d6b8129dd9d : FAILURE in 1h 14m 09s","accounts_in_message":[],"_revision_number":1},{"id":"626b4f5e8fca569c8d1696d6a81a2225ab399aa2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-04 23:25:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"caec93f3ff13e4dce1475e944656d521d59f8721","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 23:45:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0fad4112554d4143863f56e7c29390c6 : FAILURE in 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6dfc1036b2601baeb1c4346441ab9e6b635a6bc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 00:38:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc161efc0034ec2b0263acf0e162991 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bb31106fc5646979fcaa90b66722301 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eb7640322ae407d87e8fa807217b50c : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a75d1338ae54789882c496e157d7fd3 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ec706f2415f47ae85a80d103c5012af : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ed951044a8514a9eab6f8f4e38fa689f : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/55c3bd78eccc4a3093aacf60cf351135 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5d50b55810ce4869852caa8dea601b01 : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c74b6770aa9f4a52b4c4ae2a0276af20 : SUCCESS in 11m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b51954a5432d45d6abf19463efe63351 : SUCCESS in 5m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fda90b12724946c39c16e7745277d9a0 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc934779ebf043d28dd6044b28281391 : SUCCESS in 9m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f4fd1166ba4c46eaa0a09e6addb08fa6 : SUCCESS in 8m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3b2d0593348648548463875b27e52bf4 : SUCCESS in 13m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/81bc11209b364e7b9c0329793eaff721 : SUCCESS in 27m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7c3acd5d92464106ba7b2b151dc0194b : SUCCESS in 15m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/487fec3e711a4ec7ae124ae180bfb90d : SUCCESS in 17m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fe43725515d541a2be924d5432dcf7ce : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71ee8bb5e2f849f7a3fb835ac025a9c1 : SUCCESS in 6m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7480c63070a94e66bf5e00714bb19a54 : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccf94319e71e41df8cbd34e3c21f8dfc : SUCCESS in 53m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40575f987b194825955f0bd62be943e2 : FAILURE in 1h 06m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1686c88ad2913a0d9861182acc2ad4ac3829b5e3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 15:50:29.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7bfe2101da614f5e3f7553a5402cca898785d931","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 16:07:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5e7d94e2e0404d9db4f50266fc5bc30f : FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"08b3c8f8dbf79c03a69c675ff63de9381a910d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 19:24:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c097105bcd45f9be03fc9a64a19a55 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeedbca8c15f497784251b2ab46d1f66 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0c609f524ba40c4adb416f20937cf74 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8fe10f7893343f1afc9e0228ea2c0b1 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19c266290c5945fdb7d12feeb8a2e917 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56642666a292477fbe78dd8c3da1a182 : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8037e904cd474f1b962256f235ad98c0 : SUCCESS in 5m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/927360acb645404ea21b31d2f55e78d8 : SUCCESS in 11m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7859dfe25d984607989573c5441f327c : SUCCESS in 10m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9ee6e44dbd314e2d95680e5093a142dd : SUCCESS in 7m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5213fb4622e64a4da8938140998f981f : SUCCESS in 13m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/16f4475de3aa4952ae8f3cfd4b40589e : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e0c1944516754d9993c85bec83da9c58 : SUCCESS in 9m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/19aa6b3faf374b31bde4f02fcd353152 : SUCCESS in 20m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d0e73575bddf4d459d4d33a3fa532b68 : SUCCESS in 31m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dcc3350694a7497ebb9ddf35bf8065a7 : SUCCESS in 16m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bac725d3421c421395e50c9a47b9fbcb : SUCCESS in 17m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8db89085c54d4652b36544a41b64f62f : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dd82262481584e1fb85adfe694ce8043 : SUCCESS in 8m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5cc888194f2a4bb5866f17bdd60bf9d5 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa5536b64bd04086b933da54b5da709f : SUCCESS in 1h 04m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/927d64cb5ea34387a9ee6bb0f8e87539 : FAILURE in 1h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ca96a9f6755f62d0c14e5c3ce7d7765bc93c46e2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 21:18:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a45c6aeceddc51584230884f8660749a143487c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 21:41:32.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8237da1ee8ce4e17ba2cf5739eb88d4b : FAILURE in 11m 49s","accounts_in_message":[],"_revision_number":2},{"id":"604340f188dbbe946c2334abd10ec6bdebf69742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 22:13:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a3bbaaafb3451fa2abfaa499ec336b : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91982b6d3ec1418e876c700608e8e9e0 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af42dd611a084d13b0834608eb35e8e3 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/922747ce355044e490b9d85923004ff5 : SUCCESS in 8m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4cc9de18ce2740c2b06b3e03589cb0f8 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/450d4d4ba0384a76abcc210ec3c5e155 : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9860ae319fe4ba896d37b0e54e957db : SUCCESS in 7m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/63b5b1f3dd624ae1a780208126a7a459 : SUCCESS in 9m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8f0930bf09444055ba8a8143d6054044 : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1a305e0ea6044d5e9387a51afef4df0f : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/91d8faf0166a4f9cbff8f7e4eb777058 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c0f9a991341545c4b8231321e1074249 : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f52109558f184294bf622a2753d0bce5 : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b66eaafe3a4644a792ffae141d9f79cc : SUCCESS in 13m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/71900839face4ba6814ca266e83e3ba1 : SUCCESS in 27m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/db78e5c09c3a4301af42853441dcd6b6 : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c72c5d865334aeba395465c84c03c10 : SUCCESS in 18m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4079748888ef4aa9a239abb21b63ed92 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c7f6f2191e044730b6c3e7fec21ffdcc : SUCCESS in 6m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/081b40960d0541f8bb9c16d59c2181dd : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06d8322715db4f9f9cfba479cc6870aa : SUCCESS in 48m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93cb50a975634fc9aaa7be5235f14b1e : FAILURE in 49m 07s","accounts_in_message":[],"_revision_number":2},{"id":"bdfc5d9a7f370cadf21f9b66f162d604674606d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-05 23:26:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"80bb806eac55df769d2d1fa0f3932cc34b0a199c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 23:40:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cbc661ccdbb546458710ff686c450cd7 : FAILURE in 9m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e34c059fc6ab2f9b36e6052732a73135a0ab8edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 00:14:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343ea59669774d63aacf64d0873a7c0f : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cb601e06cb84e6b805f5940c03cfc3a : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/202dc1a0cad3414883e5883b97f0124a : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0172e73404724bc5bfa85cff2f271b40 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ed4a3bbca00403ea58f7ad81d7215b7 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5597c55205b24dc7a619b8330508cd04 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0d35141fb14d4f94bef7493f00c84969 : SUCCESS in 10m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c7a5a6cb4454b51b8d854e96393d0fd : SUCCESS in 7m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/df2cd263e46a4d2bab50e127e0f9eccc : SUCCESS in 11m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e0df5bf0b66641348ecfd480d2fb8526 : SUCCESS in 4m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/59aacb4f800a47e8ad828c758bee3479 : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09b70866b6134d16a17a11a87676affc : SUCCESS in 5m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fc4fb48308f74871865b7d5cbd9d242c : SUCCESS in 7m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4a67c2ccfb4040298081032e5cb44b10 : SUCCESS in 19m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/28cc9c4381634ad0b5331ce6c3e51a60 : SUCCESS in 15m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/af7c427268d44c48b1991baa28228689 : SUCCESS in 12m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/13df41f710e5456c9652194c1606adca : SUCCESS in 11m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f12003625f004b51ad210108f686ed2c : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1c03f43fb974825bd4de184f56aeb3f : SUCCESS in 5m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1fcdef337a7947899f600d7317efb6b6 : SUCCESS in 7m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de56bc69ca0f477b9258f47869b03c19 : SUCCESS in 37m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a31809ebfbdc4013a19211df07655394 : FAILURE in 42m 14s","accounts_in_message":[],"_revision_number":2},{"id":"1812014a5e1c07494798d7504e319fad162ac9b0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-13 16:03:32.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fc679469d6f5c1f026e33dd75fa28b3a7e3f6569","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 17:10:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c90f0205a4e04c0e8d4a665c60933d76 : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2f1b237d61312c60739c63f4efcbf50ceff120e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 18:32:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e4df934f3a40b68e06ba74c06544e9 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f870ec4a2e674f3ea319b71227d933f2 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a08f89d6a0de4680aebef29dcd820d11 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2aefc0c157dd4672902cb377fa86f808 : SUCCESS in 9m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/762d713da39f4b6480087cfdadb8d844 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d79bfe174584123a018b7da2798f698 : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3d331103ef7a42b99e33517705255e57 : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f581d847d2fb458ab85a44ab00ae06bb : SUCCESS in 11m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c27fa342fd44bd4988f7f8e0e730fc4 : SUCCESS in 12m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0c6d79b547204908bece7a25a15d1c36 : SUCCESS in 7m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e54f0409e5ab4b8684655dd7a884aa9f : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f1071dc29214835bbbe00e3b2cbd225 : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e95fa40f80084874b6ecd1b658b8403b : SUCCESS in 8m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/92d34533ae014d59beec954531de5233 : SUCCESS in 13m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eee2d280753441b7b6f68cbf9c2b0c11 : SUCCESS in 27m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/550099ead274429ebfef40487cc9cfc0 : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8be0483dfa0541fc96f36687770c77af : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44874bfd63b64c77a62d2e4d97ae214e : SUCCESS in 21m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f29ad18c44ac4703a297e51d78519418 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee629632b9184010a644e62edddd04bc : SUCCESS in 8m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/65fd21aa6bbc47b68bde82dfc0b212fc : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2d00106c3cf4af2b1a6308fa82a63ee : SUCCESS in 59m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fa02ad7dc564a679fdc876b1e6dd359 : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":2},{"id":"25e12ea843ee13b25b8adab8e7576023be1ade38","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-14 11:02:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e95cb7b297d95de9e164670311c8fdbab2211534","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-14 15:41:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8bc4b53b09b4ce18677433bf5156a9611a775245","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 15:42:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c36dbfdaa03377a3d9884a12509ef75c358946aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 16:56:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7c517d1d954c5f9e034fa5e05ed9b0 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e046ef46d474923b0f1c1fffafe5050 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eeca1cf630ca4c9a947d58467a6a9e9e : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37b3febe2c674003bf6160d029662bfb : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45493375be744d888833d4633a0232e5 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/989b301a4d46400b82894ceebb473eb5 : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9c0367260c147f0aad2966a413e69c0 : SUCCESS in 6m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6b15747806654903af76c5589e2c8fa6 : SUCCESS in 8m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a3b05149b9204f5c8854a46dbca59319 : SUCCESS in 8m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f61335009d784961b8e6690505cf13c2 : SUCCESS in 4m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/95e0bb9600f34aada51d76a94d0e2de3 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2b3a65d5821401c87e801aa2b7196d5 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/59db7d38c7b649fcb05f78b591672a83 : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/59392ae0c11a4accb0891aefac46b10b : SUCCESS in 16m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f22b229ba2804dcdba9c0927e34ec14f : SUCCESS in 15m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d414db120e2d469bbbdb35e13b785a1d : SUCCESS in 12m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bbbd0252c77345e497de86c4da386337 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be35009262b44d089a7ba25a8a6f3637 : SUCCESS in 15m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6eba325dfb6c45888e4f822f078170f9 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5738c6b217b046dfa5120bf3a3fb19f8 : SUCCESS in 6m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f9ab842edcb14608b03aa8660f9b997d : SUCCESS in 11m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/889a3cf9b34a424ab8b3ec118bef5085 : SUCCESS in 9m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa576a84d50945b781837a879d3ecca1 : SUCCESS in 52m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cafb4e15331464f8a507fc312626246 : SUCCESS in 1h 13m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d2612165afbcabcfb6f565e65f39b0f3f600d9ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 16:56:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2241b79d06ce1df709d71577a8f0c16092e57f85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 16:57:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae87aa767dfa4929971262164cdd4cf7 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a2aca59fbcce30fb36039ec56d467a8feba6ec7","revisions":{"ec0d9fa1b662dad36f779d6ed05fa20c5bdf7304":{"kind":"REWORK","_number":1,"created":"2020-08-04 15:27:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/744736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/744736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/744736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/744736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/744736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/744736/1"}}},"commit":{"parents":[{"commit":"ba5858c4d1c46e3da619deb6a4cb79064084b76f","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba5858c4d1c46e3da619deb6a4cb79064084b76f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-04 15:27:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-04 15:27:15.000000000","tz":0},"subject":"update constraint for os-brick to new release 3.2.1","message":"update constraint for os-brick to new release 3.2.1\n\nChange-Id: I278af4d6868273081d7579101320e0c573bd4084\nmeta:version: 3.2.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: Ic2cb11ac59e35d4b41371092ca297e9e75282bfa\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec0d9fa1b662dad36f779d6ed05fa20c5bdf7304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec0d9fa1b662dad36f779d6ed05fa20c5bdf7304"}]},"branch":"refs/heads/master"},"1a2aca59fbcce30fb36039ec56d467a8feba6ec7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-05 21:18:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/36/744736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/744736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/744736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/744736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/744736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/744736/2"}}},"commit":{"parents":[{"commit":"ba5858c4d1c46e3da619deb6a4cb79064084b76f","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba5858c4d1c46e3da619deb6a4cb79064084b76f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-04 15:27:15.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-05 21:18:06.000000000","tz":-300},"subject":"update constraint for os-brick to new release 3.2.1","message":"update constraint for os-brick to new release 3.2.1\n\nDepends-on: https://review.opendev.org/744992\n\nChange-Id: I278af4d6868273081d7579101320e0c573bd4084\nmeta:version: 3.2.1\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: Ic2cb11ac59e35d4b41371092ca297e9e75282bfa\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2aca59fbcce30fb36039ec56d467a8feba6ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1a2aca59fbcce30fb36039ec56d467a8feba6ec7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
