)]}'
{"id":"openstack%2Frequirements~744803","triplet_id":"openstack%2Frequirements~master~Idb59303aa413d8ff9b231cdea1c8b3767f13afd9","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Idb59303aa413d8ff9b231cdea1c8b3767f13afd9","subject":"Add cross testing job for os-vif","status":"MERGED","created":"2020-08-04 23:51:15.000000000","updated":"2020-08-07 02:46:50.000000000","submitted":"2020-08-07 02:44:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744803-1596768251549-ff0595dd","meta_rev_id":"c8c4cbfb77f2f7fa8b67050f8615d7edce1b0799","_number":744803,"virtual_id_number":744803,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-07 02:44:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-05 14:43:38.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-05 15:30:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-08-05 15:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-05 14:51:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-08-06 21:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-08-05 15:49:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 14:51:30.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-08-05 15:49:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-06 21:30:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-07 02:44:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7016cd8fdc8e8e81f765a8c1637160015619101f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 23:51:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ceb351a24dce14cab756cd286734abf7edc6cb49","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-04 23:54:03.000000000","message":"Patch Set 1: Code-Review+1\n\nyep looks good ill comment when i have filed the os-vif bug and fixed it. this will fail until then so you will need to add a depends on to the os-vif patch.\n\nits a trivial fix so no need to revert the upper-constraint bump for pyroute2.","accounts_in_message":[],"_revision_number":1},{"id":"e3734bae198128d8250b6957be9ddaa75af72b12","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-05 00:04:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f799c49300190975500ebfc462da20a162723dfd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-05 00:16:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"acbfb7edb71e7f5ad0bccd655c3f5b8889a6612f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 01:21:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc013778aab41f2bc0cede183152860 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e146639251e045f5b163c6629953bd8c : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e75e9abc5a344ee98f4236b79366e9fd : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e37c355d114f4e049b64acdf58c9be87 : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/94b57f96bfc647d0a7ac3b102c87ab83 : SUCCESS in 4m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dfe1d37e430245c180882f23cacd05fa : SUCCESS in 9m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c28ac1e384804794a3bdefcdbf23c538 : SUCCESS in 6m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bcb67d1b82a44ee8a86833fc2d607def : SUCCESS in 10m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f6347b14e324f27ad8c3e4a63648c26 : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/06fcf8bc0a3248798b80399ef6a88b4e : SUCCESS in 7m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4d29db7831b4452c8754373d14927ed2 : SUCCESS in 14m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa728a6b10c140acbe97a0791ae710d4 : SUCCESS in 25m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c3769a0dd95144279b712820da439904 : SUCCESS in 13m 10s\n- cross-osvif-py36 https://zuul.opendev.org/t/openstack/build/84a0594e47aa4b8784d7077c49791100 : FAILURE in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6e234d73357b40368cbae294e0fd7c5e : SUCCESS in 18m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f6d28ff6087440c0a584d5c33b4d25b1 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7593362c5132455985908980706c363c : SUCCESS in 6m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/250373e533f24b32a32f77068c086b7e : SUCCESS in 10m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc4ccce5992148d8af80cd1176bd486e : SUCCESS in 57m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6d3fd95521f44889e537166fae69c2e : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c8538fa11da8c1f8bd00fcbe6ecc380c8d0d59f8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-05 01:44:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbc8ef917dcca506ecccadeef68bb0ac4231c2b6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-05 01:44:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a427a884509b4f007ac66d06ddc0b4a1182205df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 04:00:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb82abbb4c0a4d08b781dd52e44bed0c : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5952892ea3a4d73a214cfa83290054c : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5496ce55ba9241889596ee05c85d16d3 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c7b6c589999417ba586845f64662b28 : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11f14ea75fb34d369494b4a0b004db70 : SUCCESS in 3m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0677d0a72cc74e18b540a1ac38bd1d28 : SUCCESS in 11m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fc48e3c2c00a42aa9aa39cdd4fb12244 : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d0fdafc0d4bf41e28b978abe0db6edcf : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52e7dbb584d5483b89978305e7dd5692 : SUCCESS in 9m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6a21f5d637cb4f0db0e78ea46dd6f82a : SUCCESS in 10m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e7dd354dde1848c185cc3f60f8024d67 : SUCCESS in 22m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/69c917c5b64c4f9697918e35dd32be54 : SUCCESS in 28m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/974771b0b5cf497d8eefbeece0f56077 : SUCCESS in 22m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9889bfce2a654a3c8a916ee492d980df : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22feb83d3532438f88d801433e842986 : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ed40ebcb1b18466889895ea7e4b84f87 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f17ac594d142477cafe7f82c861d0a27 : SUCCESS in 6m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f00e38546cd5455ba506d5ec725bc4f7 : SUCCESS in 12m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/902f8a23201c4075b949d75faf377e5f : SUCCESS in 2h 14m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/406652c820c84580ad15f8b90d51d230 : SUCCESS in 1h 23m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f5e2b0035047316853ab8a08c545f0aff8033d0f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-05 05:25:43.000000000","message":"Patch Set 2:\n\nwould this particular test have caught the pyroute2 issue or was it caught via func-tests?","accounts_in_message":[],"_revision_number":2},{"id":"b1c199d11360e0afc55955a3f320d19afb3101b1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-05 14:43:38.000000000","message":"Patch Set 2:\n\n\u003e would this particular test have caught the pyroute2 issue or was it\n \u003e caught via func-tests?\n\nyeah, unit test jobs failed with pyroute2 0.5.13 - https://zuul.opendev.org/t/openstack/build/f7196297e52a4aefae2679a1ce9a410e\n\nI think unit tests job should be enough for future catch also.","accounts_in_message":[],"_revision_number":2},{"id":"f3e14cb2347dcf9e6de1206591b5c916552f29ce","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-05 14:51:30.000000000","message":"Patch Set 2: Code-Review+1\n\nin this case it actully did not cause a runtime issue it just broke our tests. the code was robust enoughto check for an empty list too.\nthat is why the nova and neutron jobs which use os-vif passed\n\nbut we have been broken by requirement updates in the past although very rarely.\n\nthe functional test in os-vif require sudo and both ovs and brctl to be present so the cross job would have to be slightly different then normal\n\ni think unit test would be ok if we get broken by something in the future we can always add functional tests too.","accounts_in_message":[],"_revision_number":2},{"id":"ef5b46df97dcd76e4bc3615b4ba4de58486ec614","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-05 15:30:23.000000000","message":"Patch Set 2: Code-Review+2\n\ncool, thanks","accounts_in_message":[],"_revision_number":2},{"id":"8eb57cdff115ec8f6da57a552312062414fb3a4a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 15:49:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"70d82e15ead4304cdbfea3c1c72a2425d2781d19","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-06 21:30:41.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b49ffaf352e9b3fe81dbe14ecde1974742f063d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 21:31:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"458e74bd2265c9b37fccf6b5f1b0beb82e028783","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 02:44:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a0d0a2f5f5f0f48b507097090ce2bdb9d0c084e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 02:44:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c8e91a8f526453ab88ed5aff8c24da7 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03db8075755843feb5be49146cfeb118 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d880b7026caa4afaa30e4621d0fd17c0 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7df9984352544309ccb5f2791454ac7 : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da119be24a0f4b31aecb9f2c9f621d71 : SUCCESS in 4m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ff74d602b2454fc4ac3ac3117686b846 : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/512f3f94aeb84d3286474a0edc21a626 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2a85f4d70f6c4931be802e496a5bf7eb : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b9a5303cc4942fd93116cf57950c846 : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/377c8baf2d6f425cadfdbe887ecc337f : SUCCESS in 7m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7590439ec788405996381349abaad4ab : SUCCESS in 12m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1d553a099a714a509b166a7a2188967e : SUCCESS in 28m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a182456394b04b90b33d482e9befd8a7 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa21ce05eeee4d588f84bd97a9a292e6 : SUCCESS in 17m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/84c3afb34cb64c05b3675359b5ea4ddf : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2609cc694b394f20a6d135015f790a8e : SUCCESS in 6m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c5548731f5724266a69754e02196e877 : SUCCESS in 10m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c036d530eb6a42718c5fa7c5b57eda62 : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f873fc418a554fa5b13ae489f8df0edb : SUCCESS in 54m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a293dce138c4648a545e825925e127e : SUCCESS in 1h 30m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c8c4cbfb77f2f7fa8b67050f8615d7edce1b0799","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 02:46:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff2c575b02547008f7c5d968f866ab8 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ac1979bde2c4eadcd95951d25052d7ca9b2a7bd","revisions":{"a3b1991d271d33abdc9c86bd2f35c1a1612d21f6":{"kind":"REWORK","_number":1,"created":"2020-08-04 23:51:15.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/03/744803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/744803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/744803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/744803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/744803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/744803/1"}}},"commit":{"parents":[{"commit":"680f96b6359451287e124ff5e40f26badfe4d534","subject":"Lower upper-constraints for requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/680f96b6359451287e124ff5e40f26badfe4d534"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 23:49:43.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 23:51:14.000000000","tz":-300},"subject":"Add cross testing job for os-vif","message":"Add cross testing job for os-vif\n\nos-vif gate is broken on new pyroute2 0.5.13\nversion, let\u0027s run the cross gate job to fix\nsuch failure in advance before upper constraints\nare increased.\n\nChange-Id: Idb59303aa413d8ff9b231cdea1c8b3767f13afd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3b1991d271d33abdc9c86bd2f35c1a1612d21f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3b1991d271d33abdc9c86bd2f35c1a1612d21f6"}]},"branch":"refs/heads/master"},"5ac1979bde2c4eadcd95951d25052d7ca9b2a7bd":{"kind":"REWORK","_number":2,"created":"2020-08-05 01:44:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/03/744803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/744803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/744803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/744803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/744803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/744803/2"}}},"commit":{"parents":[{"commit":"680f96b6359451287e124ff5e40f26badfe4d534","subject":"Lower upper-constraints for requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/680f96b6359451287e124ff5e40f26badfe4d534"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-04 23:49:43.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-05 01:44:49.000000000","tz":0},"subject":"Add cross testing job for os-vif","message":"Add cross testing job for os-vif\n\nos-vif gate is broken on new pyroute2 0.5.13\nversion, let\u0027s run the cross gate job to fix\nsuch failure in advance before upper constraints\nare increased.\n\ndepends-on: https://review.opendev.org/#/c/744809\n\nChange-Id: Idb59303aa413d8ff9b231cdea1c8b3767f13afd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ac1979bde2c4eadcd95951d25052d7ca9b2a7bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5ac1979bde2c4eadcd95951d25052d7ca9b2a7bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
