)]}'
{"id":"openstack%2Frequirements~744978","triplet_id":"openstack%2Frequirements~master~I980190c4f9ce30cc01aa4c353ea976249c1112d2","project":"openstack/requirements","branch":"master","topic":"rbd-iscsi-client","hashtags":[],"change_id":"I980190c4f9ce30cc01aa4c353ea976249c1112d2","subject":"Update rbd-iscsi-client to 0.1.8","status":"MERGED","created":"2020-08-05 19:29:45.000000000","updated":"2020-08-05 22:56:21.000000000","submitted":"2020-08-05 22:54:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744978-1596668047403-ff0595dd","meta_rev_id":"1aa9bfc4d0fcd767898eec3cf10da9f55d1fc4c4","_number":744978,"virtual_id_number":744978,"owner":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-05 22:54:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-05 19:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-05 20:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-05 20:59:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-05 19:31:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-05 20:59:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-05 22:54:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"89b0531a851a7cd1432301f9a8dae091ab285723","author":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"date":"2020-08-05 19:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67dea6ad8e2f3799f0ac336446028ec82cabbaf4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-05 19:31:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"94b9259ece3e92e79bd1d800d9847c5ca95dea73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 19:54:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ed1b29cbcc5946ba86ed1b3652088bf8 : FAILURE in 14m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2236bbfacb44c113da1c906c52298d883490419c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-05 20:59:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nnormally this would be taken up by the nightly bot, but looks like it\u0027d be better to be faster here.","accounts_in_message":[],"_revision_number":1},{"id":"0c7de3ca81c81678ac0d48e303e0856b390d63cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 21:05:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb97c9baf914c6f99688f71f38c046d : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bb6ad9b0d0c4a65ab8bbf9594e6b590 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b184c52546974498bdd77fb5f00345f3 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fed89537ea3d4790bbe11c2491ba880d : SUCCESS in 5m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d49701295d7d4c409e42b6cee42c06b4 : SUCCESS in 7m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b158ac5ab14a4271990aa48aee041e05 : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6ca7b4d42ffc40ccbcb25741e53737b7 : SUCCESS in 8m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48c169d8af424ab0ad2e84df731f0226 : SUCCESS in 10m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/59c2065f7bf34be7b21985557aa35393 : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3be05de740d1447fa17a9797c8e15ef5 : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/251ed8badf024187bd9ef1b9ca7c3456 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f023a785bfd4ec1bca8c36fb96a65ee : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/217a5f70901d4626b8a1bbd26a4b8cc6 : SUCCESS in 8m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2a47316bf20b4f75965c5e389311e99e : SUCCESS in 22m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/deac65d417be439f86f3d8a96c423fee : SUCCESS in 25m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7c6ee20d6b514469a2447d6e3482ce74 : SUCCESS in 12m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7ce0b51744514656b3f73c869a601bf2 : SUCCESS in 17m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/34024b514805473bab55751925b506eb : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a2bc83029b7e47c192aefc8499bc2d5f : SUCCESS in 7m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/982bdb64e2a64c63a0831a07b6c47a29 : SUCCESS in 9m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/547b2b7def3243f7921b1198fbbbbb3a : SUCCESS in 1h 00m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3a675687a4044208989fba0c1c74f78 : SUCCESS in 1h 25m 04s","accounts_in_message":[],"_revision_number":1},{"id":"82dbc941af2f3e72f74866aab36dcaecfbe0543a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 21:06:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9ab042e115a5cbf8e89d9a87efb25a3afabf336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 22:54:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394dad218681476f893c337130fad23b : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/999d5e7336ef40eeaf28acd5aa0652fd : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70a24ba396794af5a5e5aeabb52abf42 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7e067c32aef44cfb6004a69d934a62e : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80394cc7da244b87b8331ccd4255ffd6 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/971ec035fb2249b68033fe26e0188c7d : SUCCESS in 5m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5664a60d85b646cc9939b41b47008b22 : SUCCESS in 9m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c403a37251294e30a15feae85e203c50 : SUCCESS in 11m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3d4d8313594d43ccb4de633dd21404fc : SUCCESS in 13m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57abac418e5d44f585ae681166343b3f : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c6545116cd8146daa571e3a840bd4fca : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bdc8cd22763e47bc9480198042a49648 : SUCCESS in 6m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f9b12ddf24e4487086b1193adde2e5c5 : SUCCESS in 11m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d4a34db0ad7b4be49b9ec5cc74976f72 : SUCCESS in 12m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b753bfd3c3704fd08cb9a5144aa852ee : SUCCESS in 29m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f97f43acff2149b895aab5c883dc61f5 : SUCCESS in 12m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0f4984f4ded24275b27b1c6172e67132 : SUCCESS in 22m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/987a467c1b9341f2b8ab1d8bda715a74 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c19665afcc7f4a3da1daa5618d307ce2 : SUCCESS in 9m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/99f9b212e3e540f39ba962cb942f2fcf : SUCCESS in 16m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/29d545f48bf1480eac0df996a54ea71e : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/240b9dc2cb034c919ca8f2d1f2da2d96 : SUCCESS in 1h 04m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d8cbe00464044ea8a4a84ddbe432d32 : SUCCESS in 1h 19m 45s","accounts_in_message":[],"_revision_number":1},{"id":"15612b48bcb9a094b3ad36becfa27c35c0de8354","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 22:54:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2358647046c8d15e9bfe4bd6e52c24b175561655","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 22:56:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98944c0c20c24484bfd82fdbab349f7f : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3e3f0dc37c55ad032db301dc21c90387ecfb401","revisions":{"a3e3f0dc37c55ad032db301dc21c90387ecfb401":{"kind":"REWORK","_number":1,"created":"2020-08-05 19:29:45.000000000","uploader":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"ref":"refs/changes/78/744978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/744978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/744978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/744978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/744978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/744978/1"}}},"commit":{"parents":[{"commit":"a44dfcc12233e940d7d2b9b0f853b49e83d7579b","subject":"update constraint for keystoneauth1 to new release 4.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a44dfcc12233e940d7d2b9b0f853b49e83d7579b"}]}],"author":{"name":"Hemna","email":"waboring@hemna.com","date":"2020-08-05 19:29:33.000000000","tz":-240},"committer":{"name":"Hemna","email":"waboring@hemna.com","date":"2020-08-05 19:29:44.000000000","tz":-240},"subject":"Update rbd-iscsi-client to 0.1.8","message":"Update rbd-iscsi-client to 0.1.8\n\nThis patch updates the upper-constraints.txt to\naccomodate the new version of rbd-iscsi-lient for\nthe cinder ceph-iscsi driver.\n\nThis has a new required parameter to delete_disk\nthat allows the preservation of the backing rbd image,\nto ensure cleanup of the iscsi gateway on nova detach\nwithout deleting the rbd image from the pool.\n\nChange-Id: I980190c4f9ce30cc01aa4c353ea976249c1112d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3e3f0dc37c55ad032db301dc21c90387ecfb401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3e3f0dc37c55ad032db301dc21c90387ecfb401"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
