)]}'
{"id":"openstack%2Frequirements~745070","triplet_id":"openstack%2Frequirements~master~I7b69a1df111829bb867e5c761319b239edd12dd5","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I7b69a1df111829bb867e5c761319b239edd12dd5","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-08-06 06:24:24.000000000","updated":"2020-08-06 21:23:00.000000000","submitted":"2020-08-06 21:21:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745070-1596748873207-145a67d5","meta_rev_id":"d9695e265a46eb081cd8acf06cc689f188a20dad","_number":745070,"virtual_id_number":745070,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 21:21:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-08-06 17:02:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-08-06 19:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-08-06 19:43:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 17:02:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-06 19:43:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-06 21:21:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f08bc327ec763e06818a77a83c07899d1f2e9d2f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-06 06:24:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a8263a95e764ec8de741c65abfb63aa2cc85281","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 06:59:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8a7b98d806254c099a6c507bcc2b83fb : FAILURE in 25m 34s","accounts_in_message":[],"_revision_number":1},{"id":"31b52805d7b575ef97bf40021113ae32af4322e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 08:22:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ed4a15c8ae40c886247632aca908d5 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3afa9c26a831445aba8d8bf8eb1c447a : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f4458be1590400eac619631f313fe10 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d616073a8d24411f8b0eea60be20cc5e : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14b3ecf9f9ca4e20b0ba7b58b56a00e1 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6d80f87a0d64e638dd6fd8d21066ca4 : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dfd827e24bae4f48a1327c55da86a977 : FAILURE in 6m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00e31f6b532c401991eab55dfe3f450a : FAILURE in 5m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f58ec431785f40bbb29885b12c6c31e7 : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2ea7f52b694544209670ee36b032da4b : SUCCESS in 7m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b9b612170d9d4c468fb1cd8aff05869c : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cd117eb3523745eb9a3805827c140129 : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5773420cdff84ff9bef509c618691588 : SUCCESS in 8m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4e1da82f19504a56899fe54b59aab067 : FAILURE in 15m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7a5003f7097c4bffb0b3e60376a4b9a6 : SUCCESS in 24m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c7bb057c719d40bd8ec5480024abeed2 : FAILURE in 14m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be1649ff29064a7cb57c0fb126d931e2 : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d948745e07c64b5fa3fc71e71df98b80 : SUCCESS in 6m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1562a66acff24fc7849c9ac47c133e8d : SUCCESS in 8m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6bfd3abc0ad944aba0f9ce932114c557 : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e9f91f40c564e348cc0cf958eaf9a08 : SUCCESS in 58m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e04547d91f254308b566ba20507b57c2 : FAILURE in 1h 09m 12s","accounts_in_message":[],"_revision_number":1},{"id":"17685dc4c5c16ab5cf3cefeada09fcb215abf913","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-06 16:42:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af0a9a15d27f78fcce8444cef8302f380d754500","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-06 16:42:44.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"52ab480d23b6aa081c3e786c18f499f73876a616","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-06 17:02:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a84336df6abd2939fce6dd49a559a75dfba90b86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 17:02:47.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8ce2423a81234e83aafb7645f610a57f : FAILURE in 13m 20s","accounts_in_message":[],"_revision_number":2},{"id":"59f52c03c9cde7b2fc9a0c3f55e0affed691e8ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 18:27:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bee33e7ad0554c07af870eafc0063b8f : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f799438fb8154c4db4bb68e356ae86bc : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea6110cd57be4ec4aa4c2ca954ba286b : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc467c1ecf45434681c736718314f426 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7a5372aa0a34891b08173fab4e30adc : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/110197732bd649e5aec3b605b576e733 : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/10d9bff65c0e44258b33a1910bdfe38d : SUCCESS in 7m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bf86fd1e47f64a38b73c3c54562ebb30 : SUCCESS in 9m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/62f9fa61f08b4f0193c3367c598f2bda : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/21257dc458aa4e4c85691ace0c68107a : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2b722b1492954adb984b6206de4c7700 : SUCCESS in 12m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c8bfaa06d664522ae9ecbf7b3d914d7 : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fdaa56f2282145a4a2383f52eea711ac : SUCCESS in 9m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f36148b7b427431ab312b4e41ddc5a45 : SUCCESS in 14m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/469316aeabb849ebadea2e6480d63068 : SUCCESS in 30m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8de59d45cd84407da29b105f2265b543 : SUCCESS in 13m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ea2aa7f873541ebb6b4da065edd31c7 : SUCCESS in 18m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d62817038ca94930906ace6dc778d090 : SUCCESS in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32b8f96c21754a40b94aed654940545c : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1325cdb34dcd4719b58aa7ab83c2d589 : SUCCESS in 13m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65f9986571af46d9b4148886b6dfa964 : SUCCESS in 56m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffe362b2dd684ef6885bd46d9d280e19 : SUCCESS in 1h 38m 24s","accounts_in_message":[],"_revision_number":2},{"id":"67abd8bd010ab1bf9c6221db9c12fdb8fc0dfd92","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-06 19:43:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6e890e156c5443c924f533516c8624a21b770322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 19:43:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9b4f5657d2db37ccf2c9c645d609e1c3f471e42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 21:21:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40cc7594bb5447f1a9bf89e6a547a82d : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db74d24e4dc940d898e5a636362265d4 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f49e229cd1f149598378bd47cc92693c : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2950d59b71f94d7f9a31ee9ed32286ad : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7be540d9b424bd28379390bf499b7a8 : SUCCESS in 6m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91fe0e027c004a72b8b197f495a7515c : SUCCESS in 6m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e1353962bfeb4f8f8c1977c121015c3a : SUCCESS in 11m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/367844c49ecd445fb8263901bfab0ce2 : SUCCESS in 14m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fb0ca9a36f154c84a1bde1c3c9893c4a : SUCCESS in 13m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3d638275a0cf402f9042726e20809054 : SUCCESS in 7m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e682e8aec8594c33887c59a008479f36 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/98630f208ba34a7eb7c2bce3803e634e : SUCCESS in 11m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/93a80a10025c418bba51e0a464310452 : SUCCESS in 14m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a8ddbd27af024e94a7865229404ddf7b : SUCCESS in 35m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b33d15ae7ec94c228f08591c8004cdc4 : SUCCESS in 30m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/66124b400f1c4b5793ee09367ee7e9d2 : SUCCESS in 23m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/843fc00cc75c42ebaa39eee4ed0eb7cf : SUCCESS in 21m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/939ad7fd93d34126a422a15c1216fff3 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e4b7df2fdd44257a4bce7c691ae9eb8 : SUCCESS in 8m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5b9a92e3512f4d16b2a7772ed6953bfa : SUCCESS in 22m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4d92e3eac26b40b0918abfc1de13ba92 : SUCCESS in 21m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b5563f7f4774c6f93ddd5a766b0adbb : SUCCESS in 1h 11m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ba1cd0c8518405483ecc25d528095b7 : SUCCESS in 1h 34m 55s","accounts_in_message":[],"_revision_number":2},{"id":"69ed757e918bd2a858ceae0cd983f4cf31b448cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 21:21:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d9695e265a46eb081cd8acf06cc689f188a20dad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 21:23:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7ed3862d59468b84dec934dfd2422d : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d79bc72fbe5df37e06b936eaa1af006b83c0e9b","revisions":{"3f98a589a7010cae25e8a79d491c8a147797ac54":{"kind":"REWORK","_number":1,"created":"2020-08-06 06:24:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/745070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/745070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/745070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/745070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/745070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/745070/1"}}},"commit":{"parents":[{"commit":"d0d7f3a0f1cabadd3af212e8db3a768313ad9dda","subject":"Merge \"Update rbd-iscsi-client to 0.1.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0d7f3a0f1cabadd3af212e8db3a768313ad9dda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-06 06:24:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-06 06:24:21.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b69a1df111829bb867e5c761319b239edd12dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f98a589a7010cae25e8a79d491c8a147797ac54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3f98a589a7010cae25e8a79d491c8a147797ac54"}]},"branch":"refs/heads/master"},"8d79bc72fbe5df37e06b936eaa1af006b83c0e9b":{"kind":"REWORK","_number":2,"created":"2020-08-06 16:42:38.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/70/745070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/745070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/745070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/745070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/745070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/745070/2"}}},"commit":{"parents":[{"commit":"d0d7f3a0f1cabadd3af212e8db3a768313ad9dda","subject":"Merge \"Update rbd-iscsi-client to 0.1.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0d7f3a0f1cabadd3af212e8db3a768313ad9dda"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-06 06:24:21.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-06 16:42:26.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I7b69a1df111829bb867e5c761319b239edd12dd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d79bc72fbe5df37e06b936eaa1af006b83c0e9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d79bc72fbe5df37e06b936eaa1af006b83c0e9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
