)]}'
{"id":"openstack%2Frequirements~745288","triplet_id":"openstack%2Frequirements~master~Ia50a5bad4bc01ebe540546b0bb56938f840603d9","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ia50a5bad4bc01ebe540546b0bb56938f840603d9","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-08-07 06:16:45.000000000","updated":"2020-08-07 10:22:11.000000000","submitted":"2020-08-07 10:19:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745288-1596795589345-12991b16","meta_rev_id":"12207dd3382c663393da4af00feb0e5b77e1ec8b","_number":745288,"virtual_id_number":745288,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-07 10:19:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-08-07 06:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-08-07 06:27:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-07 06:27:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-07 10:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e8bc3c8b67f7b37c716980b8fc5f230a37cebf0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-07 06:16:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6147c130bc51df6a4a2d4164d48cb2cc7de5fd3d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-07 06:22:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"463cbbbc0c585efe0d59e1bf350ce6b9b7e0f335","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-07 06:27:14.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"6990e077977bbd2c47d171fd5789a89da10ada42","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-07 06:27:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a7be12a094c968c40971491710a191c26a42a01a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 06:54:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9575fb8ab76c485f9c52679ab7771460 : FAILURE in 22m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0d5df0e4b8164ca2864cd97bbfb17c5c5ad59268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 08:24:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa478f89b2274943b358ef01df31a8f2 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bffa61d1d1646c78bed674e8f73df39 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de2afa86cd2047c2ac9f9c66a9962f5f : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d91f9ae81e9c46b49881bf805a549db4 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46da55f838cd485a87e2364e3d9a6046 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbc1b32163e44f94814e6eee96f8956d : SUCCESS in 5m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/17fbd33ab9774b34a9cb7251ab9e9a1f : SUCCESS in 8m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a231894a3f694e0d9990c8975a58e436 : SUCCESS in 10m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/59bdedb847f14c1bb5b4d34ba8c5682e : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/71914e0530224961939fdb9de2f4b779 : SUCCESS in 5m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d0dfebd4ccac4f65a46bac01d487b7f9 : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b776f343a3a408daa4ccaaf1abcf926 : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/67c35c02deb74cfb92697959678448c0 : SUCCESS in 16m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/74d4514205b949219aa96781677b44a0 : SUCCESS in 19m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6ddb6f5543524c289cfbea3052b95fc6 : SUCCESS in 25m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f79fa2a863d49adbb6c0a1a1f386374 : SUCCESS in 15m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a3107f2da96b4839bb81a861300a0b39 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/715bd218f5d44750ade0143e23917e5a : SUCCESS in 13m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aa1db05823dd495b80c569dbfacc2d9f : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/25e1d1c23173466890b1e6a60b1bd7e2 : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dc79a4d7dc6d4c66aa534abc3850e805 : SUCCESS in 18m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb72b5fccb284cb7bf6c73c188764a10 : SUCCESS in 52m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd9f8a46c3514344869d567543c01b98 : SUCCESS in 1h 10m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6794f75502d2bf30cdf209597cf4409edce08235","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 08:24:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79977702d70ddca5ab14ee41ff9b99d4e8574f97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:19:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"26a66626a2c4f7c48f69bb94bb863a0c0f2ecd89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:19:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/774b94d814fb45798268a71a466890fb : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97617ea833f34edfaea82386a80bdae3 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e00c114ad07f405abb93ae7d472b949a : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6feb01155296421098d9f71e00e32636 : SUCCESS in 4m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba4b0cb20ca4483e83ed5d27d5de93ad : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/222d14f822dc4b9893188a2a8e7b028a : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/526f1f1ea35e42ef84c3059338355861 : SUCCESS in 8m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/845f6dac49b2484bb726b75b4cce69e7 : SUCCESS in 7m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8f237aa9592842f68e6502ca7aed66f8 : SUCCESS in 9m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/275765c2289d4fe58446e8f6edb32a26 : SUCCESS in 4m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ca83bdab917247658d906d4ff76323b1 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c07e1da474bc4a8ea2f6d8ff46efc4f0 : SUCCESS in 5m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9182674c0e544194b263943ae57e4a34 : SUCCESS in 8m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3596a32704134ae4a904af3b535a0ebc : SUCCESS in 12m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8b0f9eb1dbdb46c2839aff4d3b464d3e : SUCCESS in 39m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c523e0e453174c2e9a50ad70a6cc4d68 : SUCCESS in 14m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/40d2cf72ce3647b29366bc9b4004dd4a : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90bc2ba646fe491288d4524540f27320 : SUCCESS in 18m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/20bd92cd26b149429fb4834e4074274a : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/36b127675dc6403997f7b20bf060aee1 : SUCCESS in 7m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/50a2fca4c545451d8c09eb6522eb78bd : SUCCESS in 10m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ea48e1ae3e6644e1a77fbed746e760fd : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6aa04206872c4e2cb17e3ddea231163d : SUCCESS in 59m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/616b89413b8a4d86a6392b851a1a390a : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":2},{"id":"12207dd3382c663393da4af00feb0e5b77e1ec8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:22:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15fe54bf25634684a65501e971ff057b : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"267723c3834e584bb20c96fd95699ef3db283b0f","revisions":{"2de8766cbb19a1d147daca18660340088e6d7007":{"kind":"REWORK","_number":1,"created":"2020-08-07 06:16:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/745288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/745288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/745288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/745288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/745288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/745288/1"}}},"commit":{"parents":[{"commit":"7fe36eeb5949c5854020d02f4d94a9ea4e411aa7","subject":"Merge \"Add cross testing job for os-vif\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7fe36eeb5949c5854020d02f4d94a9ea4e411aa7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-07 06:16:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-07 06:16:43.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia50a5bad4bc01ebe540546b0bb56938f840603d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de8766cbb19a1d147daca18660340088e6d7007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2de8766cbb19a1d147daca18660340088e6d7007"}]},"branch":"refs/heads/master"},"267723c3834e584bb20c96fd95699ef3db283b0f":{"kind":"REWORK","_number":2,"created":"2020-08-07 06:22:43.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/88/745288/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/745288/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/745288/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/745288/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/745288/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/745288/2"}}},"commit":{"parents":[{"commit":"7fe36eeb5949c5854020d02f4d94a9ea4e411aa7","subject":"Merge \"Add cross testing job for os-vif\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7fe36eeb5949c5854020d02f4d94a9ea4e411aa7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-07 06:16:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-08-07 06:22:31.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia50a5bad4bc01ebe540546b0bb56938f840603d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/267723c3834e584bb20c96fd95699ef3db283b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/267723c3834e584bb20c96fd95699ef3db283b0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
