)]}'
{"id":"openstack%2Frequirements~745769","triplet_id":"openstack%2Frequirements~master~Ic847052a4696bb6058aafbfffc86f039d3310c70","project":"openstack/requirements","branch":"master","topic":"bug/1891244","hashtags":[],"change_id":"Ic847052a4696bb6058aafbfffc86f039d3310c70","subject":"Revert to msgpack 0.6.1 in upper constraints","status":"ABANDONED","created":"2020-08-12 02:55:06.000000000","updated":"2020-08-12 15:32:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11d1f653a598f86cc76fd6293175231a2b064753","_number":745769,"virtual_id_number":745769,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-08-12 02:55:57.000000000","_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-12 04:12:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-12 04:12:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11ac4ee046f9dd2ea4810986bc25b4a23f668369","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-08-12 02:55:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21092044e61256b9e95b6fb09b3dba10aba212c9","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-08-12 02:55:57.000000000","message":"Patch Set 1:\n\nI proposed a fix to keystone here:\n\nhttps://review.opendev.org/745752\n\nbut in case that is not considered suitable, here is a revert of the msgpack upper constraint bump.","accounts_in_message":[],"_revision_number":1},{"id":"be2773c9c098a10f15e12d2ac9cd818c24236e58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 04:10:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d814f30f05d94de09a50a19bfdc59eef : SUCCESS in 1h 11m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d7448c9a975746e7c21f811eb54c40b490692e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 04:12:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c8df3c797041328cfbeb75bb131515 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/765c61d0a1514b608099112fa1f78d4b : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7658a81c9df846d28a114ddac723220f : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/163e8e606e434f3b82957b50f7093686 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a9d7b1abaca4d93af0b1562c250fb65 : SUCCESS in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c76824a422074407b54a0522d3725d4b : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2df42b77117a4f69a8cf28de8e7f8138 : SUCCESS in 6m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/798337f01534417bbeb50bb81116c01a : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ef3ea4e39de344cc915be81119e9316b : SUCCESS in 8m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/daca5f4773c54c2c8c616d1252213ede : SUCCESS in 4m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d42a35870f244a128320badd35de5d24 : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5120c1274b934c78b80e0de081e0fed1 : SUCCESS in 5m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2424946e71094edaa47b044a4b955e52 : SUCCESS in 5m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/813bda4cb5a440b6be2e4f51656f308e : SUCCESS in 14m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a19e88e9011b401381f6bd2c1d3b4f6c : SUCCESS in 28m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/caf30a4245ed436d80293ba6d3c9efab : SUCCESS in 13m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f292446cb6d64773b19ebe012db99257 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/feff60e0712f450e878f8458d01b48d7 : SUCCESS in 12m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8e4ce26de54b4bb0b66016ebcfaeee7e : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/339083e015bc447794422f1111d207ed : SUCCESS in 5m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/19cde972eb0b40c0aaa74eede90014bf : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfc4077edb0344d8bdde26bc50c37e90 : SUCCESS in 52m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8181a905baed42d78e2294caca2a894a : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6c2e78099f209c3f173a48663cd7a153b2fa3c96","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-08-12 15:32:57.000000000","message":"Abandoned\n\nWe are pursuing a fix in keystone for this issue:\n\nhttps://review.opendev.org/745752","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a333373248b241cc0e430b70d4201b476819bbec","revisions":{"a333373248b241cc0e430b70d4201b476819bbec":{"kind":"REWORK","_number":1,"created":"2020-08-12 02:55:06.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/69/745769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/745769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/745769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/745769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/745769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/745769/1"}}},"commit":{"parents":[{"commit":"8b919366b30a0f9bafceedffb5287ff262cbdf01","subject":"Merge \"Update PyMySQL upper constraint to 0.10.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b919366b30a0f9bafceedffb5287ff262cbdf01"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-08-12 02:51:54.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-08-12 02:54:09.000000000","tz":0},"subject":"Revert to msgpack 0.6.1 in upper constraints","message":"Revert to msgpack 0.6.1 in upper constraints\n\nChange I112d173a9858e56bcd26cbfc5beffd686fe32401 bumped upper\nconstraints and the new version of msgpack 1.0 contains a change to\nthe default data format, which is breaking *-grenade-multinode jobs.\n\nThis reverts the upper constraint bump for msgpack to unblock the gate\nwhile a proper fix can be investigated.\n\nRelated-Bug: #1891244\n\nChange-Id: Ic847052a4696bb6058aafbfffc86f039d3310c70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a333373248b241cc0e430b70d4201b476819bbec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a333373248b241cc0e430b70d4201b476819bbec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
