)]}'
{"id":"openstack%2Frequirements~746261","triplet_id":"openstack%2Frequirements~master~I08e2fb4b5d0f885ce3982f4f03c0bb1a5a51e55f","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I08e2fb4b5d0f885ce3982f4f03c0bb1a5a51e55f","subject":"Ban pecan 1.4.0","status":"MERGED","created":"2020-08-14 02:58:11.000000000","updated":"2020-08-21 13:58:57.000000000","submitted":"2020-08-14 16:32:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746261-1597422769353-3cab5452","meta_rev_id":"5aae270fc0e8b1dbaaa171aee828b875a4149a39","_number":746261,"virtual_id_number":746261,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-14 04:06:57.000000000","_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-08-14 13:55:31.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-14 16:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-08-14 12:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2020-08-14 09:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-08-14 05:29:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-08-14 12:41:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 04:06:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-08-14 05:29:50.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-08-14 09:27:22.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-08-14 12:41:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-08-14 16:32:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9695d7b987209ccc4968951a5a7f94293265eed","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-08-14 02:58:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c431e7e0bc3e3e7878a7eecb997189a01521d8f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-14 03:11:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4e30a8894751b3000f41de35119823170f5da67a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 03:23:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d2605815a8714023ab492dd7a89d2015 : RETRY_LIMIT in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"334cffb4451860dee015d2d1427d41140c3e4bca","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-08-14 03:58:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"10d015f25d9a804572e079616510500777ebfcbd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-14 04:06:57.000000000","message":"Patch Set 2:\n\nis there an upstream bug for this?","accounts_in_message":[],"_revision_number":2},{"id":"a7569398f20e93c39e373470e3a68428ded8b6f8","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-08-14 05:02:16.000000000","message":"Patch Set 2:\n\n\u003e is there an upstream bug for this?\n\nclick the task in commit message for the story.","accounts_in_message":[],"_revision_number":2},{"id":"8069b64960f4ca6f9773c87ed9a04fe218f11787","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 05:09:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/83cff40d79cf466c817dfbfbdbd90c21 : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b3ed0ac0e0dbf6d903cd22d5ed8c55823abaeba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 05:27:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a7d86966494a59a9d6a19942dcfae9 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9400f8279f434338a6f39d023ce41530 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d852c3ea3ba04d64aea02481294d1e29 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94dfe29454984fc3bc119624169b1c69 : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6175b299526643b89b655da623f95b6a : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1a5a1241efc148178e336ea4ff73ea46 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fae3c33fd5b646d3a87f28ab91b6c450 : SUCCESS in 6m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c8121776b2b4350b65014810b9dba6f : SUCCESS in 8m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/27cddbdb3b2c48d2ac5f0bc3d62de7fb : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6072af427fd648be8fb955cdda9871d3 : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a8f1a46214a249119680bd60e1ee1123 : SUCCESS in 9m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9beb24016a6f41b7a6b4db4501ab60cd : SUCCESS in 6m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a0e9943cd1154e8885c672f4d0301fcb : SUCCESS in 7m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b8a7a796549e4fb9911fc4fb7211db39 : SUCCESS in 14m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2697ee01447b4c1f9f74965284f9eae1 : SUCCESS in 22m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5d59c95285254fe492e0a2ef4ece06cb : SUCCESS in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b25426c356c494e917ba0c1537d28f9 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87a84d643f8b47bda3a1c9ae777f48f1 : TIMED_OUT in 41m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ae09f71b90354165b7b511cfff8cc407 : SUCCESS in 4m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61c114809bfc40d28a81d5fcaa76d239 : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9ac5f8190a4942769148a05b4545aca6 : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/320bd93f32d342aa92016be0a42939a5 : SUCCESS in 51m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/409ac64863ba4acfb3416131f01212bd : SUCCESS in 1h 24m 01s","accounts_in_message":[],"_revision_number":2},{"id":"cf68bd330b3c5010387c513f3ca7e67620135ac8","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-08-14 05:29:50.000000000","message":"Patch Set 2: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c716d6c0a643c570add666a57cb0e465185faed5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 06:40:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bbae20b4ef114233b6a8c81532fdb9e1 : SUCCESS in 1h 06m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d5dbb195ebab65218fa74ef4687d67eaa845c328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 06:57:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4903188a78f94ba2be5bc152b9350589 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47c2af14c487445c8e513a9fca2730c2 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3327a27b2247417ea2bf14103382fb1a : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75f1794bb7ff450da111dfc7dedf62dc : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/453d4b19138146ac98ff6d2c0d4d1972 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae0fef496e9a4721b00a20747d5843f0 : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a43e8e19c4634e5c933ad0552c8b9532 : SUCCESS in 11m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ebb98554918a4e8a9ce790db57e4d667 : SUCCESS in 10m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/29f6ad5aefca4930961c969090f74ffd : SUCCESS in 8m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/769dce9e24f9455aae9b6fd271826d3c : SUCCESS in 7m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/51f0f6cd8dcc4ad28ab924aa43b6d623 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6392d23ea2a3448cb8c07c5a396e0339 : SUCCESS in 7m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/35bc17db008c45f8a4f4648a730e9dfb : SUCCESS in 12m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4a4b5c7680f9422a836d46975473c771 : SUCCESS in 38m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b07750bc1c4d4fb891eb800805341441 : SUCCESS in 22m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/193372def0cf44f79487d1a946af07d8 : SUCCESS in 19m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/497d79d684e8492cafad77f79a043be3 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/913847640f834363abe7744d82dc1d1d : SUCCESS in 20m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/861f976e61e14583bc10c811de220125 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/178e1dcc46654cb783281357d8ed02f2 : SUCCESS in 7m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0619d160553947c5b6e0a691c4abb047 : SUCCESS in 18m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e450fcd4a07e41bd8060b2964151f725 : SUCCESS in 1h 02m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa0b3f1b2d9745eb92eeee5fdc84aaeb : SUCCESS in 1h 25m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5a67e3c51921fd96d83dd2eb692b6b48383bb0ee","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-14 09:27:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f9ca46764f5c5f6b6f85e623ed5b126cb0d4eb70","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-14 12:41:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nApproving to unblock gates, but we should make sure the issue is reported so we don\u0027t need to keep this capped.","accounts_in_message":[],"_revision_number":2},{"id":"abaf3fb7c0513bad9bbde0098d13580c888685b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 12:42:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ee47deb2e651408cbd2ee2f34413210e3ec10b2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 13:52:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea820a458c124634b0c5f7e6dbd3efa6 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25311034a2e04ad6966bac3d9ba75034 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d79f558af5ee4381831693803f09a829 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7795856df70e44349416993e8852f56e : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7976ea3d4ae14ca6a2811f0bec6f4516 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1fb09038d399491c9c6c3d86f1e31d83 : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e47d1223a88f4d6389ab2890c5eaf38d : SUCCESS in 8m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/88feeedacf7143f38fc3ee9ecc819b9a : SUCCESS in 13m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a1626334d48047cbab96ed4a08ca5936 : SUCCESS in 8m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4c1bd518087e46689c7102b3bf4c1c65 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8d1dadf805474308bef3b365bf9556da : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20e049b903be4fb2a8a58372b852c840 : SUCCESS in 6m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/308d391e57f14deb950c2bd8ab8082c9 : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c28b116e11b145059d66b255fb21a6cf : SUCCESS in 12m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ffb334501fbf4acc811a38ef17f7134b : SUCCESS in 29m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3db7673de1c143e99abf30f0e320da9a : SUCCESS in 13m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b62c1f7aff464182ac9377e18c1bf99c : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ff3b82970b64712af1b2a7657abd7e8 : SUCCESS in 19m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/46cb10d0863d41849ff260a6cfbd9f84 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80a6f952f0e64f5fa3b71a59a657534e : SUCCESS in 6m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0bb5a1fb027c4df79309b3261e5c8b11 : SUCCESS in 15m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/38632eedf11b4c2fa72c607567205244 : SUCCESS in 9m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72aa0dd22cb4412e8eef4ea20dc90737 : SUCCESS in 52m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b4247b59e7341f78d9071d34f214bdc : FAILURE in 1h 04m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d81693f4bb7dfa896612eb285cc5bb3f9717cdeb","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-08-14 13:55:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"153ad5aa8f888c8c08a18e4b44662cc5487ecb34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 15:10:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb102be201f4be1a5ec3670b2f0a4c6 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f986c06470e14548b11503d9614f60bd : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dc451da1e8d4dfa8fa6149f52829090 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0d4ad77a6b54edb9e8cef64af5f8f63 : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e87a210d2fe43bbb60f862dd0c4aad4 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fcb23cd4c9474ad5bd96ab0062c960d3 : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8f8b059deee241db835963ea68294f6a : SUCCESS in 6m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a911daf53d1441fb8446d4b852ffb1f1 : SUCCESS in 10m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9458ca5456694288ad0e42118b280483 : SUCCESS in 13m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/04856df75d7a4f3cabc017a494d2b708 : SUCCESS in 5m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a279164de9a14b49a7aa1327f5e97289 : SUCCESS in 12m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/873c349bdc1741d28b44a7f8a41a5c84 : SUCCESS in 7m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a6681afa57464e4d9d29a6a384d4add8 : SUCCESS in 9m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fb1edcae1e35460bb1c17b50cee69234 : SUCCESS in 20m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2fbab0113c064df9bd3618647ca4b79c : SUCCESS in 31m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cc076574eb4d4c6f8b777f5c092bf92a : SUCCESS in 17m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0497c32a1059448f93bee8912a569e8b : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/44667357682743f6a8e8fd66106ed220 : SUCCESS in 24m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3b4a87d2f76d4cee96da22b8796378e0 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91fa253f465c4d25a5fce210fea7b648 : SUCCESS in 7m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6f3b2d934c7141ba86ff7039049a3a4f : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45c84450077540c281cfc71606d6cc66 : SUCCESS in 55m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8241cfb762b44611adccfbcf49280745 : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d718e79a773fccf6c74aa880ee7c6d1375121e53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 15:10:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/966b26b389f146e2a701c0937b46de44 : SUCCESS in 1h 11m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e9b6c603c4a98a52170c2c419432399824b95203","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 15:10:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c79fc9ff732216c489b11e2179d9b8c235209e24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 16:32:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce5c4ebb9adb4f90b40b734c13fd14fe : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8e98b427d740c887c6582ae9d1caab : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25729a8f02f640dba3b1d2f246576ab2 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7764f4975bdf48758c62503f3ae8b3aa : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d49a67b06d074f77941b9c2fc759ecb1 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5495de4ac1c445f3a511027df868302c : SUCCESS in 3m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/91cbf56db00e4fadb781d2ce567da4f0 : SUCCESS in 7m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f956d98427714222a7e84126ccca1083 : SUCCESS in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b4a9aa57d7f344cc8aaa65a8055156f2 : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/301509cf111747889841b8ead2f5e56a : SUCCESS in 5m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d1daabf3a34842c6b0231ec3974e5dba : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/008fd03d530c4a109f2e8ad959039f49 : SUCCESS in 7m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f598eb286b80401cbe69acd721aa84c5 : SUCCESS in 8m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ee8554f8c3e943b9be78e96f9e4bf695 : SUCCESS in 13m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66e90489f13a49858ccacec5a46297e7 : SUCCESS in 21m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/02c76bbb4a9c4256bc334c0a0b42133e : SUCCESS in 16m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3763fc868432413ebfdbdffb37618649 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/744ab15c6e43404ab7cdc95b0bb99f6b : SUCCESS in 19m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e27db3bcd5d0455cb300270f5d21bf31 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/971e3b6bb295414bb0997e7719baa208 : SUCCESS in 6m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac2cc70c3da5411395303252e74c9add : SUCCESS in 15m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/995cc7e011724120a10d7d34bb6c1985 : SUCCESS in 16m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a11f2bdc3bb464fac50d3f099126a19 : SUCCESS in 53m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a664fd3abe204956a4e14a096a5dcf43 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1ce6454692f4e0dad846c9f07b384373524a5313","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 16:32:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e5afd8198d16025214604e8dcc9fe0721a791fa4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 16:34:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb5a1fcf5fa44c8949d414598ebd080 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2},{"id":"9c0b08ab31d9e86002293d61817bcbc8a4791621","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2020-08-15 09:42:35.000000000","message":"Patch Set 2:\n\nI\u0027ve created a python-gnocchiclient issues and submitted PR.\n\nhttps://github.com/gnocchixyz/python-gnocchiclient/issues/109","accounts_in_message":[],"_revision_number":2},{"id":"5aae270fc0e8b1dbaaa171aee828b875a4149a39","author":{"_account_id":5792,"name":"Dan Radez","email":"dradez@redhat.com","username":"dradez"},"date":"2020-08-21 13:58:57.000000000","message":"Created a revert of this change as Id5759fcc7e856c2acab8f0f4214017e7d344c0f1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"570307ea8f569fa73445b745325e9245dd7b42b0","revisions":{"04c1c3ca1a05e3d4ddf37eb213d325842223ca73":{"kind":"REWORK","_number":1,"created":"2020-08-14 02:58:11.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/61/746261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/746261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/746261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/746261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/746261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/746261/1"}}},"commit":{"parents":[{"commit":"9e480271ddf286860aec118cbcf34d553c3e637b","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e480271ddf286860aec118cbcf34d553c3e637b"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2020-08-14 02:56:42.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2020-08-14 02:58:07.000000000","tz":330},"subject":"DNM Ban pecan 1.4.0","message":"DNM Ban pecan 1.4.0\n\nChange-Id: I08e2fb4b5d0f885ce3982f4f03c0bb1a5a51e55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04c1c3ca1a05e3d4ddf37eb213d325842223ca73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/04c1c3ca1a05e3d4ddf37eb213d325842223ca73"}]},"branch":"refs/heads/master"},"570307ea8f569fa73445b745325e9245dd7b42b0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-14 03:58:24.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/61/746261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/746261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/746261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/746261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/746261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/746261/2"}}},"commit":{"parents":[{"commit":"9e480271ddf286860aec118cbcf34d553c3e637b","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e480271ddf286860aec118cbcf34d553c3e637b"}]}],"author":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2020-08-14 02:56:42.000000000","tz":330},"committer":{"name":"Rabi Mishra","email":"ramishra@redhat.com","date":"2020-08-14 03:57:53.000000000","tz":330},"subject":"Ban pecan 1.4.0","message":"Ban pecan 1.4.0\n\nThis has broken gnocchi which in turn breaks ceilometer\nthat uses gnocchi as the default backend. This has also\nimpacted heat gate which uses gnocchi based alarms with\naodh.\n\nPlease check https://review.opendev.org/746034 that\ndepends on this to fix the ceilometer gate.\n\nTask: 40652\nChange-Id: I08e2fb4b5d0f885ce3982f4f03c0bb1a5a51e55f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/570307ea8f569fa73445b745325e9245dd7b42b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/570307ea8f569fa73445b745325e9245dd7b42b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
